)]}'
{"id":"openstack%2Ffreezer-web-ui~986971","triplet_id":"openstack%2Ffreezer-web-ui~master~If5617f1914051839bdfd25de91b01889cc8b79b7","project":"openstack/freezer-web-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-01 13:23:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-01 13:25:37.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"If5617f1914051839bdfd25de91b01889cc8b79b7","subject":"Declare Python 3.14.","status":"MERGED","created":"2026-05-02 11:09:47.000000000","updated":"2026-09-01 13:38:23.000000000","submitted":"2026-09-01 13:37:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986971","meta_rev_id":"df2bb421b6e4053598f7ff81eba0dfa41302b8e2","_number":986971,"virtual_id_number":986971,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 13:37:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-09-01 13:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-01 13:23:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-02 11:44:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 11:07:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"f4bd19b5f9e4c72a59ca30408fbf6337ca67c42b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 11:09:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"061574e629a0b6af95f038f331d235053c0e13fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 11:10:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6143c322084af9fb2ce776819d8781c2c3445701","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-02 11:10:27.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"f22ecdf8343e45dcdab34a246191658846aa6342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 11:44:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94714225637c4ec6ba10b3e676696ce5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5125b4835d9c4407bb8030fe41c03850 : SUCCESS in 3m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/46150dc99a934c8aab7dc10ca74b7e53 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8547bb0a3b5849c39516ac39bc284cbd : SUCCESS in 2m 52s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/3254aedd4bff47c98280fb55a87fb1ab : SUCCESS in 3m 21s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/4037b5f451af4a3e904b04c50b9426d2 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbaeb5c5223c4416bf964ab0f1fb6403 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf098dff4c0e4d9e95ebd16e5f59cc3f : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6fc8df03a904565bfb6a05d89ab114d : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8182dcb5c5d749eb9ab7768413422a01 : SUCCESS in 5m 43s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/ef193d16e40e4cf1af502239fc5470a0 : SUCCESS in 28m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a3af99ab18cf4e279f658b4077c71467 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"36d4d01aacf31b1e07e68f1b089143978edd4946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 16:36:18.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ba869aebd06adc7b998d062d1020d2163cfdfdb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 16:36:37.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c3184a045879cab7ddc0cd4e29a32a3e6455d866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 16:50:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f41ac563bec34ac2a386a59d9fcd865a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85821e56101d42c887670bf6575f6603 : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2b4fb0a2a754b30a0f700a628e2db6c : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7bbd78d908c4e619c33a2837ca4522e : SUCCESS in 3m 41s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/27b1bc3e59b54c099bffb6e6f394e7c9 : SUCCESS in 3m 53s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/e342278113eb47498cbb18347761407d : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3385080d2720489c951ec001b66fc79d : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/447460b7017b47c69c5dca39f70dfd49 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db12cdb7bdda4fbb92bf7f739500694e : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fc8f0342e148268685da77eaf8d2bf : SUCCESS in 7m 56s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/1bcd7f143cb44ffd9a37b78a01f2e204 : SUCCESS in 12m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5fa9ecc28b8e45b4ba51523e6577a841 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":4},{"id":"6836b056a2e81a1cf33b791c401eaa7b8af88972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 01:57:53.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9f76f1a62e0958cc1fe3128a7f8ff8073a49cb88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 01:58:12.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"89fe78cad974737f6be3e27ae3d70a3131d9e2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 03:04:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098a1b7485f54a3cb9e59a01f4bb462c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a184d1fdfb454a87bc35624bdf4b47d8 : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4627912683634961a57d227234fd0a0a : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47940e1782a746ebafc44dbdd8142139 : SUCCESS in 4m 15s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/aa13417db73640e3b2c9ae499b5a9550 : SUCCESS in 6m 30s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/a8c44971f6aa4d27af46ff5d4ace0a40 : SUCCESS in 3m 48s\n- horizon-tox-python3-django60 https://zuul.opendev.org/t/openstack/build/fe0eabd9111d4534a5e1a8e26f3f07e0 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb0c02e8db3d4c9caf33bb815306503c : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c386225dcbb423e9b02eca35d710cab : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a62846e6b60a4852839ec0c59dd41dbb : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a674e5ecef8f4ccfb7285ec6d36b9b35 : SUCCESS in 7m 57s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/00e916d6e37d4a00803356e53943aa00 : SUCCESS in 27m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de45fa3897eb4e109738c6fadbca4141 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":6},{"id":"4e2aff28bbe585026275025ce07c5aa4fea79d00","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 13:23:05.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5e5b0ca8d488fedc94f3aed3d7ec6c0d84ed2488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:23:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5f367a33434965e43ae42bcf9b56d20440cd93c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:37:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c55ca2a26d74464cbd6a67622bb77961\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6b4cc4998734ca79776d70a7f57a114 : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d603cb3fec1415dbdf742e1c9d2cc1e : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26ac973c73424529ad237456d3d938de : SUCCESS in 5m 40s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c8b5607fb7a44e88a639b3520298e793 : SUCCESS in 3m 31s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/77b2182100874005869f705a46d4fc06 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5094ebe5518452fb188473743a9efa7 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c63d6b8295b44783824c30534cbc7271 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4267abcee7e5462e9df4e71196c548c9 : SUCCESS in 9m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ced83278d13a421b9a6792694ead19e8 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"e6aa2f87051f63cb0897a5b90a3220c8b0a9556b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:37:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"df2bb421b6e4053598f7ff81eba0dfa41302b8e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:38:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/266110458d674efb94e53e290031e6ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8bfc762a02b4d07b8e0647274b08618 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3b45159fc9d43f963548565fa31a5ce3c91de170","revisions":{"0e4c98a83aa73e1aa0752c46f65b8425489240ec":{"kind":"REWORK","_number":1,"created":"2026-05-02 11:09:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/986971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/71/986971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/1"}}},"commit":{"parents":[{"commit":"77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41","subject":"Merge \"OpenStack Authentication Version - remove old versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:09:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:09:47.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus).\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: If5617f1914051839bdfd25de91b01889cc8b79b7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/0e4c98a83aa73e1aa0752c46f65b8425489240ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/0e4c98a83aa73e1aa0752c46f65b8425489240ec"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"25c4ed5b1b26fea83bc99deccdf9722f03fba7e3":{"kind":"REWORK","_number":2,"created":"2026-05-02 11:10:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/986971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/71/986971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/2"}}},"commit":{"parents":[{"commit":"77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41","subject":"Merge \"OpenStack Authentication Version - remove old versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:09:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:10:24.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus).\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: If5617f1914051839bdfd25de91b01889cc8b79b7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/25c4ed5b1b26fea83bc99deccdf9722f03fba7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/25c4ed5b1b26fea83bc99deccdf9722f03fba7e3"}]},"branch":"refs/heads/master"},"3ab51b6794bfcff0cf92660440b244521c7ef04a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-05-04 16:36:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/986971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/71/986971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/3"}}},"commit":{"parents":[{"commit":"77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41","subject":"Merge \"OpenStack Authentication Version - remove old versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:09:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 16:36:18.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus) and\nadd Python 3.14.\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: If5617f1914051839bdfd25de91b01889cc8b79b7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3ab51b6794bfcff0cf92660440b244521c7ef04a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3ab51b6794bfcff0cf92660440b244521c7ef04a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2ea25d50b72e35c6dc6b491f171d1a4c6cee8fed":{"kind":"REWORK","_number":4,"created":"2026-05-04 16:36:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/986971/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/71/986971/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/4"}}},"commit":{"parents":[{"commit":"77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41","subject":"Merge \"OpenStack Authentication Version - remove old versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:09:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 16:36:37.000000000","tz":0},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus) and\nadd Python 3.14.\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: If5617f1914051839bdfd25de91b01889cc8b79b7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/2ea25d50b72e35c6dc6b491f171d1a4c6cee8fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/2ea25d50b72e35c6dc6b491f171d1a4c6cee8fed"}]},"branch":"refs/heads/master"},"d29667d328ea5d1288a3befe8868c17c32131aa9":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-09-01 01:57:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/986971/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/71/986971/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/5"}}},"commit":{"parents":[{"commit":"77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41","subject":"Merge \"OpenStack Authentication Version - remove old versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:09:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:57:53.000000000","tz":0},"subject":"Declare Python 3.14.","message":"Declare Python 3.14.\n\nChange-Id: If5617f1914051839bdfd25de91b01889cc8b79b7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/d29667d328ea5d1288a3befe8868c17c32131aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/d29667d328ea5d1288a3befe8868c17c32131aa9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3b45159fc9d43f963548565fa31a5ce3c91de170":{"kind":"REWORK","_number":6,"created":"2026-09-01 01:58:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/71/986971/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/71/986971/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/71/986971/6"}}},"commit":{"parents":[{"commit":"77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41","subject":"Merge \"OpenStack Authentication Version - remove old versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/77a0bdb4d2ce5d8f97ac099cd78e7e213c3b6a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-02 11:09:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 01:58:12.000000000","tz":0},"subject":"Declare Python 3.14.","message":"Declare Python 3.14.\n\nChange-Id: If5617f1914051839bdfd25de91b01889cc8b79b7\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3b45159fc9d43f963548565fa31a5ce3c91de170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3b45159fc9d43f963548565fa31a5ce3c91de170"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
