)]}'
{"id":"openstack%2Ffreezer-web-ui~993607","triplet_id":"openstack%2Ffreezer-web-ui~master~Id48e2f75d52f72b7cce14bbd9ae2818edfa57aa1","project":"openstack/freezer-web-ui","branch":"master","hashtags":[],"change_id":"Id48e2f75d52f72b7cce14bbd9ae2818edfa57aa1","subject":"Align jobs/sessions dashboard layout","status":"MERGED","created":"2026-06-16 14:26:22.000000000","updated":"2026-06-16 18:12:47.000000000","submitted":"2026-06-16 18:09:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993607","meta_rev_id":"3e03009a80ab60d13c6c1d997b1e5943fa458e5a","_number":993607,"virtual_id_number":993607,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 18:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-16 17:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-16 17:49:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 15:50:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3984aecdf5edf2c07182bcedf2434e6f2823b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 14:26:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7b19d43343c0f58a72160b8de82acaeacfcac76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:50:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d0aa1af82764fab9cc87a81f1a4effb\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab2c4e4a03ea4c8385031ba4d8c4c465 : SUCCESS in 2m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d1907bf7912443e698958d7206110106 : SUCCESS in 6m 22s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/f0726601bef24ea68fea71be5dfa25bd : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c2ff24cf2243bc8235e881c70f41fc : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66794f87f247408ba5295af1461c0c40 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/100ad6e6c1024f37a8af5af6a2a0d0e9 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53c19a1a4904fed88a6fd9e5214f5eb : SUCCESS in 8m 21s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/1ad79434f62b40fcbe162d2b4dd009a5 : SUCCESS in 32m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6188ec3378a5444c91b3bce11defcfb5 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6ac05681d44ff6db7b019d4534efc2896673dbb1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 17:49:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f700bf15228285cea99267b1619e2dddaf6e4ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:50:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1c9c772b578a2f9f23e7dc2b075516b6dd13d3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:09:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604d04ee37264622bfea1949b6a9f433\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4d7c2817b8a4fb59574bb6fb74c9d9d : SUCCESS in 2m 42s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e5fa98b0a24f4b63ad1690e491400e79 : SUCCESS in 4m 24s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/421a3dd0a07842888c54f05bbb2d919b : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508d09a74f2046aaaba67eb483eb0608 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4edfe24e81e45e8a556c42c6706d6b1 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4501bf9066124b339d2706df3c164f26 : SUCCESS in 8m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01960164d30a4ccd8bc4db4dad39e0fa : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1bea02aae9fc1a701ccf8d9f45a807b4c25260f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:09:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e03009a80ab60d13c6c1d997b1e5943fa458e5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:12:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04764195a46c4202b1e7b1fb5c280520\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed4d0299bfc460790d6be114b8fc4df : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2344b1dd6d5f27af59bd6f1140f71f1a1c12279","revisions":{"e2344b1dd6d5f27af59bd6f1140f71f1a1c12279":{"kind":"REWORK","_number":1,"created":"2026-06-16 14:26:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/07/993607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/07/993607/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/07/993607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/07/993607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/07/993607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/07/993607/1"}}},"commit":{"parents":[{"commit":"a58e5dfb1bc6fa849f16cc21c092f76041faf939","subject":"Merge \"Add interval unit/value selector to session creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a58e5dfb1bc6fa849f16cc21c092f76041faf939"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 14:26:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 14:26:13.000000000","tz":120},"subject":"Align jobs/sessions dashboard layout","message":"Align jobs/sessions dashboard layout\n\nJobs and Sessions are attempted to be shown on the single screen,\nwhich makes an impression of broken dashboard as a whole,\nas this concept is quite unconventional in Horizon.\nMoreover, it does not help much, as looking for details still take a\nwhile and reloads the screen.\n\nSo let\u0027s transition it to a classic view of horizon\ndashboard where selecting the specific item would open a separate page.\n\nChange-Id: Id48e2f75d52f72b7cce14bbd9ae2818edfa57aa1\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e2344b1dd6d5f27af59bd6f1140f71f1a1c12279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e2344b1dd6d5f27af59bd6f1140f71f1a1c12279"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
