)]}'
{"id":"openstack%2Ffreezer-web-ui~993653","triplet_id":"openstack%2Ffreezer-web-ui~master~I5b1be26bd7ef53e3713bf1ed2ecfb67c2291234d","project":"openstack/freezer-web-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-17 12:32:51.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I5b1be26bd7ef53e3713bf1ed2ecfb67c2291234d","subject":"Refactor Action assignment to Job","status":"MERGED","created":"2026-06-16 18:27:36.000000000","updated":"2026-06-17 12:48:33.000000000","submitted":"2026-06-17 12:47:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993653","meta_rev_id":"fd625e4740c6094da1b9e78e33484ce3bb3c7ce6","_number":993653,"virtual_id_number":993653,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-17 12:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-17 12:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-17 12:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 18:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7384af62077609cfc41f2d41a6cedf1eaac361ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 18:27:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96593feaebdeec0b83ebc6f86f44b4c93028b6b2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 18:29:43.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"a24f4011f66e7147cbf4a85da0b10e7a8ee2fb5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 18:29:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0475e75fb00fc2f277e1c9af74859bab3d436fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:45:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7147cd8da1be4e8ebc59580c98682e03\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adfea48fbe854fdbabfe245a794ceff6 : SUCCESS in 4m 01s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a1f5799ddbdf49509adcf6da4b1b9f50 : SUCCESS in 5m 35s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/46676a930dc74a1185ca6352a4bffadc : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c921402daec0433a8f670f3efa2cbb4d : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f92770c88d6f44d6bb615430010d4ae2 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd9468012bff48a88ab9e770458750dc : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98393e235864b8882c2ca2c17d67816 : SUCCESS in 4m 25s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/344ad2278b5f4b64a3cc171cee47d9a4 : SUCCESS in 12m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b738005ac4594dcab1f428518aa870fe : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f01a6222f7557fda49c002d181f05391331bde1c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-16 18:53:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cab6ba4b9e1811b1a1e4e259bf13f71779d19eae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-17 09:35:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"326057b3451b62a1a9a20af334b9166de63aa633","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-17 09:35:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e82567c15cbcc1f13f69e348450d42f983878c2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 09:35:10.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/freezer-web-ui for 993653,3","accounts_in_message":[],"_revision_number":3},{"id":"16e273c8a424542274000525864f424805d99a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 10:08:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e6118c30be346b4bcde5614743d6b6e\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cdaa39e40f3407380b73b9fb89880aa : SUCCESS in 4m 11s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/26aefd50c13e430aac4f488541cf2b85 : SUCCESS in 4m 49s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/54b561eff12c421fa737f2d96dc7c2da : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2876262f9574c949572aa7d2844b99e : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0c8a91ac3e148b3bf8a5923085af82d : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38d442b3c27c4a089d1e940645e46605 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265f4bdc5e47432a93dd683421314bba : SUCCESS in 7m 26s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/97d7eb1fb95341f686c98bc15a3c0c31 : SUCCESS in 28m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1390bef4c13e4f3082775fe86bad870d : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":4},{"id":"82215e50e13cd3cbfcca5c546f623e0466a79d71","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-17 12:27:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"e91c1d89ee03d6661c1b7ba207792d8fcaaba21e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-17 12:32:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5c9af13ae1dd4465ee190fa1834ed5df5affd2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 12:33:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"34516c9402b9ff7ccf49fa2c51bc2cb6001feacb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 12:47:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c631339a824b66ab817c3e823279a8\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5971dfcd922b457d8f8204487670ed66 : SUCCESS in 3m 25s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ce3c9031d2fb4bae82bd8e98cce8b2b9 : SUCCESS in 3m 49s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ff20a38d2f974045b125fcf6cdf06389 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f895f2a5444d4828a0d1127063b139ee : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a13a3a364b6c44d1a67ee63e9d133d70 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5d9908c7dc4c948331820c159764ea : SUCCESS in 7m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/260f1b5b12ad4c2fbe6d419665d3743f : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":4},{"id":"8189fe13bba2e49a95f6573dca2665fe2891e459","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 12:47:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fd625e4740c6094da1b9e78e33484ce3bb3c7ce6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 12:48:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bfa9d0b098e4ae7afa40538b1db2756\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8621658f68f4dc1a948ae83d44ab0d4 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f6c6f2e6395ed7d8c97bd427bfde480b0cc2f4bd","revisions":{"fd56ce10d53efc29d9dbd3eefa75a61a5489b8bb":{"kind":"REWORK","_number":1,"created":"2026-06-16 18:27:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/993653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/53/993653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/1"}}},"commit":{"parents":[{"commit":"3db44c8c43398fe2224739300cb1c8bdee6c1dda","subject":"Do not show delete button for central agents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3db44c8c43398fe2224739300cb1c8bdee6c1dda"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 18:27:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 18:27:25.000000000","tz":120},"subject":"Fix Action assignment to Job","message":"Fix Action assignment to Job\n\nIn Job actions there is a possibility to attach or detach an action from\nthe job. This behavior was broken, resulting in impossible user flow\nto create a functional job.\n\nAssisted-By: Gemini 3.5\nChange-Id: I5b1be26bd7ef53e3713bf1ed2ecfb67c2291234d\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/fd56ce10d53efc29d9dbd3eefa75a61a5489b8bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/fd56ce10d53efc29d9dbd3eefa75a61a5489b8bb"}]},"branch":"refs/heads/master"},"3ecc38dd4048eeed60059b9feaf987adb726a5d5":{"kind":"NO_CHANGE","_number":2,"created":"2026-06-16 18:29:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/993653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/53/993653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/2"}}},"commit":{"parents":[{"commit":"35da8ff07ae0344b9993d0b643d23c8112167967","subject":"Do not show delete button for central agents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/35da8ff07ae0344b9993d0b643d23c8112167967"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 18:27:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 18:29:54.000000000","tz":0},"subject":"Fix Action assignment to Job","message":"Fix Action assignment to Job\n\nIn Job actions there is a possibility to attach or detach an action from\nthe job. This behavior was broken, resulting in impossible user flow\nto create a functional job.\n\nAssisted-By: Gemini 3.5\nChange-Id: I5b1be26bd7ef53e3713bf1ed2ecfb67c2291234d\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3ecc38dd4048eeed60059b9feaf987adb726a5d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3ecc38dd4048eeed60059b9feaf987adb726a5d5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3db44c8c43398fe2224739300cb1c8bdee6c1dda","ours":"fd56ce10d53efc29d9dbd3eefa75a61a5489b8bb","theirs":"35da8ff07ae0344b9993d0b643d23c8112167967","merge_strategy":"recursive","contains_conflicts":false}},"a2aba50fdb2dd6b59b050d7c032b02080fba2f6c":{"kind":"REWORK","_number":3,"created":"2026-06-17 09:35:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/993653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/53/993653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/3"}}},"commit":{"parents":[{"commit":"3db44c8c43398fe2224739300cb1c8bdee6c1dda","subject":"Do not show delete button for central agents","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/3db44c8c43398fe2224739300cb1c8bdee6c1dda"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 18:27:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-17 09:27:25.000000000","tz":120},"subject":"Refactor Action assignment to Job","message":"Refactor Action assignment to Job\n\nCurrent AJAX implementation of Actions assignment to the job relies on\nthe obsolete sync-xhr functionality, which is widely prohibited by\nweb-servers and proxies due to creating locks.\nIn order to simplify the flow, we move the functionality from the\nfrontend to backend, which removes usage of xhr for this action.\n\nAssisted-By: Gemini 3.5\nChange-Id: I5b1be26bd7ef53e3713bf1ed2ecfb67c2291234d\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a2aba50fdb2dd6b59b050d7c032b02080fba2f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a2aba50fdb2dd6b59b050d7c032b02080fba2f6c"}]},"branch":"refs/heads/master"},"f6c6f2e6395ed7d8c97bd427bfde480b0cc2f4bd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-17 09:35:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/993653/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/53/993653/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/53/993653/4"}}},"commit":{"parents":[{"commit":"b2c745f6b25f50a0daceca84bedcbcce527d555d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b2c745f6b25f50a0daceca84bedcbcce527d555d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-16 18:27:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-17 09:35:10.000000000","tz":0},"subject":"Refactor Action assignment to Job","message":"Refactor Action assignment to Job\n\nCurrent AJAX implementation of Actions assignment to the job relies on\nthe obsolete sync-xhr functionality, which is widely prohibited by\nweb-servers and proxies due to creating locks.\nIn order to simplify the flow, we move the functionality from the\nfrontend to backend, which removes usage of xhr for this action.\n\nAssisted-By: Gemini 3.5\nChange-Id: I5b1be26bd7ef53e3713bf1ed2ecfb67c2291234d\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f6c6f2e6395ed7d8c97bd427bfde480b0cc2f4bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f6c6f2e6395ed7d8c97bd427bfde480b0cc2f4bd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3db44c8c43398fe2224739300cb1c8bdee6c1dda","ours":"a2aba50fdb2dd6b59b050d7c032b02080fba2f6c","theirs":"b2c745f6b25f50a0daceca84bedcbcce527d555d","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
