)]}'
{"id":"openstack%2Ffreezer-web-ui~993929","triplet_id":"openstack%2Ffreezer-web-ui~master~Ifdc3c7642c4a18dc807529e3eb863c02ff4e9dc7","project":"openstack/freezer-web-ui","branch":"master","hashtags":[],"change_id":"Ifdc3c7642c4a18dc807529e3eb863c02ff4e9dc7","subject":"Verify ts is valid before conversion","status":"MERGED","created":"2026-06-18 14:26:58.000000000","updated":"2026-06-18 15:21:08.000000000","submitted":"2026-06-18 15:19:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993929","meta_rev_id":"1e47e01d700def0ce69c599e0059624f64302937","_number":993929,"virtual_id_number":993929,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-18 15:19:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-18 14:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-18 14:33:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 15:05:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62cdb7562d2f88f81cb58fc59633dd8aa987efa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-18 14:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"567aad1799d8b296c7c2ebeafb9cc811b0e98ed8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-18 14:33:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"008ce01b925b22e418936bd0e6c72169e4080c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:05:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc2e1689c62426bade4bb81eaca93d3\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92112aa449294502aa59d043f8301ea2 : SUCCESS in 2m 16s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7b423cdaa4eb445f8ded79ad70ba5fed : SUCCESS in 3m 01s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/63482b380964423cb019a9c5e608419c : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97210155233c43089edbbd7618fc11a5 : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2e79190fcc94370b3c2821278e99505 : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/727efa6681e94823b0d24885118575de : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb67f974bd04b49bd9642fe299af86c : SUCCESS in 4m 38s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/91da95722a604b95b2ce9feef8bfe3e5 : SUCCESS in 32m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/999720d788be4986859eb03d98c621aa : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e631db5cd2f0e73c8ba3e36063bde16c5cb1ed7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:06:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f1246ee85d23f78603509da0b45c8160078accf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:19:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aafe791906b47d5972fc978ffe84a50\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e38a788a9d84456b918ea2c2fbc09ab6 : SUCCESS in 4m 12s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ecffe6def9f248089aa96ed6698062d6 : SUCCESS in 4m 43s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/c9ab5b647b2a4fa9962ed1e3325aef75 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd32905a54dc4ed48d4606f9f8faf9b6 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e370fa316104376baabf7c5df8b3947 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74861c505da4a659d15ab9c7c7ab8fe : SUCCESS in 8m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8293b5b7d4d541bda6bfb305cd92d9a6 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5bb042873b204ed9c48c449b6d5790fad8ffd0a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:19:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1e47e01d700def0ce69c599e0059624f64302937","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:21:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa62c249d5d4ec482ec06892a8cc2aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0f9b2dd65b49838879e352bf5c0d66 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ad603d095f04d0e19e2c3ce3930d263d2f97e8f","revisions":{"7ad603d095f04d0e19e2c3ce3930d263d2f97e8f":{"kind":"REWORK","_number":1,"created":"2026-06-18 14:26:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/993929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/29/993929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/29/993929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/29/993929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/29/993929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/29/993929/1"}}},"commit":{"parents":[{"commit":"f6c6f2e6395ed7d8c97bd427bfde480b0cc2f4bd","subject":"Refactor Action assignment to Job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f6c6f2e6395ed7d8c97bd427bfde480b0cc2f4bd"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 14:26:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 14:26:49.000000000","tz":120},"subject":"Verify ts is valid before conversion","message":"Verify ts is valid before conversion\n\nFreezer API tends to return timestamp as `-1` value when it can not\nparse it from the database or it is simply not present.\n\nSo instead of returning Dec 1969, we let template to use it\u0027s default\nvalue for such cases.\n\nChange-Id: Ifdc3c7642c4a18dc807529e3eb863c02ff4e9dc7\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/7ad603d095f04d0e19e2c3ce3930d263d2f97e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/7ad603d095f04d0e19e2c3ce3930d263d2f97e8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
