)]}'
{"id":"openstack%2Ffreezer-web-ui~993970","triplet_id":"openstack%2Ffreezer-web-ui~master~I0453fb97608fd50b216004d95c5c65319a494de3","project":"openstack/freezer-web-ui","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-19 13:18:59.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-24 09:06:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0453fb97608fd50b216004d95c5c65319a494de3","subject":"Remove external load of JS files","status":"MERGED","created":"2026-06-18 16:54:24.000000000","updated":"2026-06-24 09:08:50.000000000","submitted":"2026-06-24 09:06:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993970","meta_rev_id":"02b048743289ad15544671e76415132b596ef9a0","_number":993970,"virtual_id_number":993970,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 09:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-24 08:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-24 08:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 17:09:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-19 11:24:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"193deb4078a3c68d4de08da046a0f76b1536c337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-18 16:54:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30824617192e544228ee4f8160803ea6e0a1555d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 17:09:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba82199644c4022a53d486e1ebade02\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d27b365ea39e45a39f59f51647524d52 : SUCCESS in 3m 44s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/49af11b6e0364ee5b75d196cf07b7a18 : SUCCESS in 5m 27s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/41e8d5e84405406b87268cd5bb15e6a1 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e978078a5f4ecfb7561a4345ce00a5 : SUCCESS in 2m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/427bd484d0e749c2af36a10aa2d2d201 : SUCCESS in 2m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9b2f3fe613e46bd97dcd96d99e2e526 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfd1feea74a4ca79e2bde05300652f6 : SUCCESS in 4m 19s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/5957a43e083942088ea69f688c6cb36d : SUCCESS in 13m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e647df2d5b8a44c187068359f4176115 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8e5083fd45c7e0882c775056b862ab5f88dab86b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-18 17:39:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a1b4506acc9db2108aec770b21926cb57ac21796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 18:21:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa480ede03447f5b490bdf6b82e9cb5\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec6b1554209d433998d59ae918576d4e : SUCCESS in 3m 47s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/45590a00a3b549baa86cf43fe5183c8f : SUCCESS in 4m 42s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/1dfc7cbfda364b0eaddf9bcb0f230c24 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3602604bdeea43529064bd544238a734 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d532719424d8448692e63e7d1b2da93a : SUCCESS in 2m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8754a61517044c1bbdd12c285a70327 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47b8db79abe413c8f026cd888e29f1f : SUCCESS in 7m 50s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/60999ea2290a40b695a678459abfa477 : SUCCESS in 35m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d366ae23f854e66b6407074e26283a5 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"14441c5935f6e90d23425607ce1d3303e6831928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-19 11:12:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f5ea947077dfd04d402ed9fbb93bc6ca123d5e34","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-19 11:24:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"30a7ddba20edcfe36f6268b7c8410e68866b33b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 11:44:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a7a0805a3534bd8b5c3101fbdaf4191\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fede2ba1833049ca90a9d1468cbb950c : SUCCESS in 3m 40s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ca71157290ff4395be69287911460c19 : SUCCESS in 6m 06s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/36b139f8f65b4f2ab82c68d678f73c6f : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec1dec8f74049208fdc9696ce9d55a0 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/162d9079d824416d90707c03a0959709 : SUCCESS in 2m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/232a74490a624179bb7d02668d2fdece : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b7f5c1f9a64295b04f762973ba9c69 : SUCCESS in 8m 37s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/9635c8502e1943f8a0048d4f6c039089 : SUCCESS in 26m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93667b3afb6a41e9972e3fcf5949f92f : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"9a690a2ecf3f57d8160fdd1d6ae89aac754bb902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-19 13:18:21.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b3fc5eaf82cf505fdf62fe5b2a30912b864162d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-19 13:18:59.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"70de8a28b9550c8d9b97300000fecda153af6694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 14:03:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b9300db4d648879d68771f2b0be3f8\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76d52af83f194066bcc76bfc8897584a : SUCCESS in 2m 44s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5225c082e38b48998da9ecb8cd7aeab7 : SUCCESS in 5m 02s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/7a2e66a0dba9466eb9971ce510f2453e : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e1cb1f1d4e5476dbf21f6ba166b3d25 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c440cd8145834da79bfc1ed08404c062 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/60a8aa0ee33a4680afdbaf2f5cc9cd80 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0a6edc563e40668979fb694826eafc : SUCCESS in 8m 57s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/c27fa775e52943fba3cd9d3a0623fb3a : SUCCESS in 15m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/253ad18ccbdd4536b618cf80d555db76 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"18e12bc62e2f53fa68964c7cbc998a678a0bc9e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-24 08:50:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e6ae3bce82cc92a3c63d0f03ca71d84d56fe3f4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 08:50:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a28baba2d3c6aa8ca193c01676b221426072ab64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:06:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c746ba7d1cc4c939a89499db99a48a7\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a7da6f1cd244fd2bc76be85950ec3ce : SUCCESS in 4m 05s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/10014da32c7141a2a16d048b0ad38a85 : SUCCESS in 6m 28s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/021413408cf84e5688231902a9c27789 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3922fa7661524429a926d186aaf155aa : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f156c77f7daa40a0aa660bd56081da52 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff5cc2f93724eb481ae4257544e98b8 : SUCCESS in 9m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38a09f2d848846ecbd76e134744c309b : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"c93787f808cee1f660cc418d2264ba391a2d08e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:06:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"02b048743289ad15544671e76415132b596ef9a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:08:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21df922c865e44cc8e79b0d236b998f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f325d07c19df4b8a91ef19bc6e4c887a : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f3ade871c4d6c02778eb30189a52d7203ca4965f","revisions":{"e968bc6e1b6a9cb1b3b2982770bebf86bf7271a1":{"kind":"REWORK","_number":1,"created":"2026-06-18 16:54:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/993970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/70/993970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/1"}}},"commit":{"parents":[{"commit":"1397e4a4a32cfdd633952fd49c8ac4a50d46028a","subject":"Merge \"Verify ts is valid before conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1397e4a4a32cfdd633952fd49c8ac4a50d46028a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 16:54:16.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 16:54:16.000000000","tz":120},"subject":"Remove external load of JS files","message":"Remove external load of JS files\n\nWith collecting/compressing static files there is no need for explicit\nincludes.\n\nMoreover, as these files are collected and compressed, we need to ensure\nthey do not always exucte and refactor JS to run only when they\u0027re in a\ncorrect context.\n\nChange-Id: I0453fb97608fd50b216004d95c5c65319a494de3\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e968bc6e1b6a9cb1b3b2982770bebf86bf7271a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e968bc6e1b6a9cb1b3b2982770bebf86bf7271a1"}]},"branch":"refs/heads/master"},"064ada37f7a12f5ef47250aaa1ee7732ba0776c2":{"kind":"REWORK","_number":2,"created":"2026-06-18 17:39:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/993970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/70/993970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/2"}}},"commit":{"parents":[{"commit":"1397e4a4a32cfdd633952fd49c8ac4a50d46028a","subject":"Merge \"Verify ts is valid before conversion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/1397e4a4a32cfdd633952fd49c8ac4a50d46028a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 16:54:16.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 17:39:43.000000000","tz":120},"subject":"Remove external load of JS files","message":"Remove external load of JS files\n\nWith collecting/compressing static files there is no need for explicit\nincludes.\n\nMoreover, as these files are collected and compressed, we need to ensure\nthey do not always exucte and refactor JS to run only when they\u0027re in a\ncorrect context.\n\nChange-Id: I0453fb97608fd50b216004d95c5c65319a494de3\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/064ada37f7a12f5ef47250aaa1ee7732ba0776c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/064ada37f7a12f5ef47250aaa1ee7732ba0776c2"}]},"branch":"refs/heads/master"},"101069a57a00eed14e9381ac3ecb696c373e8e86":{"kind":"REWORK","_number":3,"created":"2026-06-19 11:12:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/993970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/70/993970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/3"}}},"commit":{"parents":[{"commit":"b0b8bcb2b7fdaad6b88ae3e57989601553302cb5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b0b8bcb2b7fdaad6b88ae3e57989601553302cb5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 16:54:16.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-19 10:42:34.000000000","tz":120},"subject":"Remove external load of JS files","message":"Remove external load of JS files\n\nWith collecting/compressing static files there is no need for explicit\nincludes.\n\nMoreover, as these files are collected and compressed, we need to ensure\nthey do not always exucte and refactor JS to run only when they\u0027re in a\ncorrect context.\n\nChange-Id: I0453fb97608fd50b216004d95c5c65319a494de3\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/101069a57a00eed14e9381ac3ecb696c373e8e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/101069a57a00eed14e9381ac3ecb696c373e8e86"}]},"branch":"refs/heads/master"},"f3ade871c4d6c02778eb30189a52d7203ca4965f":{"kind":"REWORK","_number":4,"created":"2026-06-19 13:18:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/993970/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/70/993970/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/70/993970/4"}}},"commit":{"parents":[{"commit":"b0b8bcb2b7fdaad6b88ae3e57989601553302cb5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/b0b8bcb2b7fdaad6b88ae3e57989601553302cb5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-18 16:54:16.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-19 13:18:07.000000000","tz":0},"subject":"Remove external load of JS files","message":"Remove external load of JS files\n\nWith collecting/compressing static files there is no need for explicit\nincludes.\n\nMoreover, as these files are collected and compressed, we need to ensure\nthey do not always exucte and refactor JS to run only when they\u0027re in a\ncorrect context.\n\nChange-Id: I0453fb97608fd50b216004d95c5c65319a494de3\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f3ade871c4d6c02778eb30189a52d7203ca4965f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/f3ade871c4d6c02778eb30189a52d7203ca4965f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
