)]}'
{"id":"openstack%2Ffreezer-web-ui~998633","triplet_id":"openstack%2Ffreezer-web-ui~master~I8dbb940dc8d886b69d5ad022a4cde13b193f7656","project":"openstack/freezer-web-ui","branch":"master","topic":"freezer/backups_update","attention_set":{},"removed_from_attention_set":{"39044":{"account":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"last_update":"2026-08-18 10:44:53.000000000","reason":"removed on reply"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-18 11:01:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I8dbb940dc8d886b69d5ad022a4cde13b193f7656","subject":"Display backup status","status":"MERGED","created":"2026-07-24 15:32:40.000000000","updated":"2026-08-18 11:14:48.000000000","submitted":"2026-08-18 11:13:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998633-freezer/backups_update","meta_rev_id":"b1584413261e3cd41b828c2234bc1098b4fdc388","_number":998633,"virtual_id_number":998633,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-18 11:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"all":[{"value":1,"date":"2026-08-18 10:44:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":2,"date":"2026-08-18 11:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":1,"date":"2026-08-18 11:01:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-24 16:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-27 11:34:58.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"CC"},{"updated":"2026-08-18 10:44:53.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"REVIEWER"}],"messages":[{"id":"78f57ad02ddcd03fb37ec9a481b51f7f3d379e3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-24 15:32:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7768d144e604fa45e79888f17cb5b71abca5cb2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 16:14:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77af5309284d4df79987b03482bc126f\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4715b69cdef8416a9d2a0ef31d946067 : SUCCESS in 3m 33s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/709f2ee9b8ea4ffab4ea853681c8dc56 : SUCCESS in 4m 29s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/89a95b2aa56e4394a5b23630303ba5ff : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114d5bec7f0448ceb3accbb4f403df0c : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a04f27ab78a0447ba4a744a83214f4f1 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e284d8813b2c442583f771279dc6b2ca : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f13dba4db142d89a3e35ab19be0192 : SUCCESS in 9m 24s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/e4ba6e471e194d1fb80511c1f93d8dd5 : SUCCESS in 34m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/704e4cbbe7874a75a19cb497d82a5fcf : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8ba6c9decff0821199d51e30bc2264d254782747","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-07-27 11:34:58.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"346bc62730545f27510c7d3c83829138557d4825","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-12 06:04:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30e131b00e414b1dd1854d51cd25fdcc6b5d7904","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-12 06:05:04.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"af8a7163d35a545817111bbbd9bd8662b7eea570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 06:05:08.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"340d91e5f48017a9d3604845c37c679ec4926117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-15 15:33:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_28619\u003e\n\nOutdated Votes:\n* Verified-2 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":2},{"id":"e0c08d9e79b4cefe3c893d3a5607e78d027ca026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-15 16:07:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f9b0a542a9420983c875a8c7a603c7\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bebed0bc6414eddb2dbddda458501a1 : SUCCESS in 2m 32s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/9f11b21fd98546fe91ae5ea718e85d7e : SUCCESS in 3m 04s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ff9497ef64254d59a0ffa869eba6d139 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5157e4ab0244a26ac2e28d1668f4525 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/185ea7030043463b8f5f08fc5dc5b532 : SUCCESS in 2m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b20f607528504699aa226bce1d63b1b6 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31195e029eb44e94b0f5a090a9954a02 : SUCCESS in 4m 26s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/cf0c959f924c4b29ba06f216fef9cbe1 : SUCCESS in 28m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86e1446b0d4045bfb9a926f007b03c0d : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"06d317f5e1f26dae54f6b9350b24d40ac373eaf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-17 14:19:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"911181d38830aade86c372568c9b8fefd0b37cea","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-17 14:20:17.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"519318bfbfadf09ad44d0c4c00c22c9826cd054d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 14:58:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/110213cc8f924fd7a67276f4dfff908a\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/318f459d0bfc47b0a0b8f96160b6ea96 : SUCCESS in 2m 48s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5901b20eee9c4fab906e003117b61e9d : SUCCESS in 4m 29s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/bd7352e44d2b415a86d7271f5ac9e72d : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed3798e7dc64ddb87fbeb252cd6de2e : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47c89164548f42ac80e928b0c4cb9246 : SUCCESS in 6m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e0e6232ee50c498c8e2e45c4fc4d6afc : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f049cd028d444212b90dfcaf23bce83c : SUCCESS in 5m 48s\n- freezer-web-ui https://zuul.opendev.org/t/openstack/build/68d3a52ccd19477ab39d7a4a3c4957a9 : SUCCESS in 26m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5f70e0d3b5840acb7daf65791884bb6 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"a5f217af6ebe17b4c6e75abcd37a2d12dbfd8bcd","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-08-18 10:44:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a9c01807774a07bb2288deba7d92586889ad3151","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-18 11:01:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"23964c5249c3b5a906abf7859569f52279e96944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:01:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0bcaa1a05e0f87169da8072d513266e40d3b8420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:13:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5be9aca1987476fa52e851222ee29b6\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/168eccea8ae6409ebe821257953aa6d7 : SUCCESS in 5m 06s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b15f9dab2bab44f3aa4af88184be153d : SUCCESS in 4m 20s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/bd11977254dd45e1801ac1dc1efb5c3b : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac9522c07d040e4a9d58284b27bc811 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7aed6078fa9248efac4f53cd0ca563ce : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6205a57dbfda45c9bbccb639111ca736 : SUCCESS in 7m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fb83da3594154ccc888724812898c826 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"ce47399df1627f7b4c8a871e4ed56c589da134f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:13:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b1584413261e3cd41b828c2234bc1098b4fdc388","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:14:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bae403e5e14c41089e5ebf713f4c751c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d264b3b078400a997c4428b91cde4e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e0fdb74f4458555ca2e440a2ca58e92e3ecd38c2","revisions":{"a0b8887c07d843f1cdd4e14bf70ad694ace0737f":{"kind":"REWORK","_number":1,"created":"2026-07-24 15:32:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/33/998633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/33/998633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/1"}}},"commit":{"parents":[{"commit":"08177e0d41652e3dffa61bead290b1b62a174c12","subject":"Add job management into session create/update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/08177e0d41652e3dffa61bead290b1b62a174c12"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 15:32:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 15:32:32.000000000","tz":120},"subject":"Display backup status","message":"Display backup status\n\nWith introduction of backup status in the API, let\u0027s ensure it\u0027s also\nshown up in Horizon.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/998519\nChange-Id: I8dbb940dc8d886b69d5ad022a4cde13b193f7656\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a0b8887c07d843f1cdd4e14bf70ad694ace0737f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/a0b8887c07d843f1cdd4e14bf70ad694ace0737f"}]},"branch":"refs/heads/master"},"2c565d1ee6e4eea9ee5e370c75becfd4a50a1ef1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-15 15:33:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/998633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/33/998633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/2"}}},"commit":{"parents":[{"commit":"8b328305bec7218044e4003a60e49e4f8b269a5c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/8b328305bec7218044e4003a60e49e4f8b269a5c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 15:32:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-15 15:33:49.000000000","tz":0},"subject":"Display backup status","message":"Display backup status\n\nWith introduction of backup status in the API, let\u0027s ensure it\u0027s also\nshown up in Horizon.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/998519\nChange-Id: I8dbb940dc8d886b69d5ad022a4cde13b193f7656\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/2c565d1ee6e4eea9ee5e370c75becfd4a50a1ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/2c565d1ee6e4eea9ee5e370c75becfd4a50a1ef1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"08177e0d41652e3dffa61bead290b1b62a174c12","ours":"a0b8887c07d843f1cdd4e14bf70ad694ace0737f","theirs":"8b328305bec7218044e4003a60e49e4f8b269a5c","merge_strategy":"recursive","contains_conflicts":false}},"e0fdb74f4458555ca2e440a2ca58e92e3ecd38c2":{"kind":"REWORK","_number":3,"created":"2026-08-17 14:19:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/33/998633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer-web-ui","ref":"refs/changes/33/998633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer-web-ui refs/changes/33/998633/3"}}},"commit":{"parents":[{"commit":"8b328305bec7218044e4003a60e49e4f8b269a5c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/8b328305bec7218044e4003a60e49e4f8b269a5c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-07-24 15:32:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-17 14:19:04.000000000","tz":120},"subject":"Display backup status","message":"Display backup status\n\nWith introduction of backup status in the API, let\u0027s ensure it\u0027s also\nshown up in Horizon.\n\nDepends-On: https://review.opendev.org/c/openstack/freezer-api/+/998519\nChange-Id: I8dbb940dc8d886b69d5ad022a4cde13b193f7656\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e0fdb74f4458555ca2e440a2ca58e92e3ecd38c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer-web-ui/commit/e0fdb74f4458555ca2e440a2ca58e92e3ecd38c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
