)]}'
{"id":"openstack%2Ffreezer~1001496","triplet_id":"openstack%2Ffreezer~master~I846c246323a0b9f6e6b90f5de2217402bcc8cc90","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-20 14:10:06.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I846c246323a0b9f6e6b90f5de2217402bcc8cc90","subject":"Validate job before creating clients","status":"MERGED","created":"2026-08-19 13:56:10.000000000","updated":"2026-08-20 15:06:26.000000000","submitted":"2026-08-20 15:03:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001496","meta_rev_id":"d3c50cd8267f684464e7604a4d002749498ed140","_number":1001496,"virtual_id_number":1001496,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-20 15:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"all":[{"value":1,"date":"2026-08-20 10:49:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":2,"date":"2026-08-20 14:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":1,"date":"2026-08-20 14:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-19 14:56:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-20 10:49:03.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"REVIEWER"}],"messages":[{"id":"41b7dae0f348d007b9aa1bac47016ad4ea99fcb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-19 13:56:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a63327a05660796f0c668567a4b7df63a9eddab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-19 14:06:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8d2a0aa7cde85a8be7b49fee7eb17c3f3b0c6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 14:56:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e36adfb737ed45b397e4cf35fe7086f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85952a755cd94441a9414c0f7268eb09 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5c514abe041492fa42592d405d3b3f5 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87fa7a674434d309ca76c5fa3c5d103 : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e44f7514f8e4b328c0b8e23ed512a83 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1cd458cd37847a494d31bf073da6499 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4b222f8fc3b4abbad5edba029cb63d2 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b5b65d995244836905c6e7fd673d476 : SUCCESS in 3m 59s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/fd41297fd8ca491b9216291bd72d0371 : SUCCESS in 19m 30s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e321ab876c834d92850e5ae69902d9a6 : SUCCESS in 31m 20s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aba488877bea4910ab39eaabae37ac74 : SUCCESS in 34m 34s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/e5b8ab1552c7419ea0b3ab6774bca6d7 : SUCCESS in 43m 59s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f43e609e02d4fb19cd5844dcc978675 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6eedf2338ce10d7250bc2f1a6ecc8487e763cad0","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-08-20 10:49:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5fa48da837aa17a52f7723aff4c08156374a5d11","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-08-20 10:49:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7f216d5d27263b0bcb9db48a94bb51b20f5339f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-20 14:10:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4c313e04a6869f60f90bec04f8cc7a5dbc1e34f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 14:10:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2ea2158502369647fd6409d4dbefb9c7e22ce25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:03:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef473da2aac442b28dff2640cf250820\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89c5c25249d4d409a2379a077029fcc : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82bb485da37645ee98c002af1d2362be : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/589a2eaf305a4783864a9bb1e0b8a08d : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38589af4e28749e4bf87496710fb702d : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c0aa6a007094e9384616f6eada38dad : SUCCESS in 4m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8f99ca522a84498807eb0b9e34023ba : SUCCESS in 4m 56s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ac91f9b866824a1698db58b4631f8772 : SUCCESS in 33m 44s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/75b05e24bfd6408893ee736a86f67b8e : SUCCESS in 40m 30s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec360d01af474358bc19ae004b90d2b0 : SUCCESS in 37m 22s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/515395f5da37424d9e69287472e97e6e : SUCCESS in 47m 46s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/d56f257f2fcf49e4a2aba77dc6ff4409 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":2},{"id":"396aa317128e7f4367fab88a336d59f81a2e05ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:03:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d3c50cd8267f684464e7604a4d002749498ed140","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:06:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/496da6a0264f4d97858fc5a37b2e41e4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8967bed26f482795b1b78f86a15c63 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c938ff5b2264ce07011bbad907c84d7efde01c81","revisions":{"7e3ba35287c77829f5a6f03b588c9db1930a5999":{"kind":"REWORK","_number":1,"created":"2026-08-19 13:56:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/1001496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/96/1001496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/96/1001496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/96/1001496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/96/1001496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/96/1001496/1"}}},"commit":{"parents":[{"commit":"6dbf34d1e1f63e1e9fc61e98b4b47f1edc61f41f","subject":"Merge \"[doc] Refine service documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6dbf34d1e1f63e1e9fc61e98b4b47f1edc61f41f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-19 13:55:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-19 13:56:08.000000000","tz":120},"subject":"Validate job before creating clients","message":"Validate job before creating clients\n\nIn case job is invalid, there is no need to make external communications\nand create clients.\nSo all client creation should be coming after validation is successfull.\n\nChange-Id: I846c246323a0b9f6e6b90f5de2217402bcc8cc90\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7e3ba35287c77829f5a6f03b588c9db1930a5999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7e3ba35287c77829f5a6f03b588c9db1930a5999"}]},"branch":"refs/heads/master"},"c938ff5b2264ce07011bbad907c84d7efde01c81":{"kind":"REWORK","_number":2,"created":"2026-08-19 14:06:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/1001496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/96/1001496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/96/1001496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/96/1001496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/96/1001496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/96/1001496/2"}}},"commit":{"parents":[{"commit":"6dbf34d1e1f63e1e9fc61e98b4b47f1edc61f41f","subject":"Merge \"[doc] Refine service documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6dbf34d1e1f63e1e9fc61e98b4b47f1edc61f41f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-19 13:55:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-19 14:06:50.000000000","tz":120},"subject":"Validate job before creating clients","message":"Validate job before creating clients\n\nIn case job is invalid, there is no need to make external communications\nand create clients.\nSo all client creation should be coming after validation is successfull.\n\nChange-Id: I846c246323a0b9f6e6b90f5de2217402bcc8cc90\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c938ff5b2264ce07011bbad907c84d7efde01c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c938ff5b2264ce07011bbad907c84d7efde01c81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
