)]}'
{"id":"openstack%2Ffreezer~1002760","triplet_id":"openstack%2Ffreezer~master~I80eb5d1a63d6d2caf150665642871ec1268cab8f","project":"openstack/freezer","branch":"master","topic":"fix/1599592","attention_set":{},"removed_from_attention_set":{"39044":{"account":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"last_update":"2026-08-28 15:35:10.000000000","reason":"Change was marked work in progress"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-28 14:31:04.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I80eb5d1a63d6d2caf150665642871ec1268cab8f","subject":"Properly pass hostname and container name on pre-create","status":"ABANDONED","created":"2026-08-28 13:51:19.000000000","updated":"2026-09-04 11:20:04.000000000","total_comment_count":3,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"1aba83e8f488580f35780ab963ad76459fb8b189","_number":1002760,"virtual_id_number":1002760,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-28 17:06:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-08-28 14:27:03.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"CC"},{"updated":"2026-08-28 14:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14cc2f32b32d671b716fe565c2b683a079f69258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 13:51:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2809d77f40e6a471e59f4868cfb68fc5b89a1a48","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-08-28 14:27:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ba82325c3b34fbf54ee07cb49d4869db42d6af0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 14:31:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"369c4422c08908b3f1063e978cd1a74785cb87b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 14:38:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc866894a864150b73d2b2bf68c0839\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5383df476545bc8933e56151ab65ea : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9d4fc1c9ded47b49205075d6cdd2889 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651b41e2bb524bc8b5d9c9448736760b : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eab33f5a6a434f13b8912d888b635fcc : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afde36b0df3a43aab3af75518ea86b84 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff6acc27a3504289bd41b4845857fe33 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e1f4c31385542238fda4c4ec3de9d65 : SUCCESS in 4m 44s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/377caeda5c41460fb2f49e0b1d862729 : SUCCESS in 36m 05s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/5b1e65c7e61c4e5db27ddc58d6cd75c4 : SUCCESS in 41m 10s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8766b443fd1c40ce9abcb4396685319a : SUCCESS in 37m 07s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/bc0a46708c6c458db967c64ae28d520e : SUCCESS in 23m 24s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/1bb63071c7bb42d7bdae15caf64ffcbb : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"444a6dee79bf8c41a1c76ba1369ce547b300425f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 15:35:10.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5054dc11536d02625207c106418675fc3a023f01","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 15:35:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d8ce902946afed6a8d3ccec457cedf41d9339e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 16:23:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bd5474f876b8c1e693eae473f87c862c5839e33a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 17:06:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f069add6aef94eb197c3a79751fbbd3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ba8be75c124d708b4cce054e9c003f : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119ea0e1d46a437bb88d38ced6c4dfef : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e88f1219d54112ba123eb832df303d : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04388d88fb154c4f96c5c9a88264e08e : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad37e8988a7d47e9aaf68612499fc23b : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2bd01616ec144b759bda869070f2888d : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c6e8d9fb2fb46a8a83252c090cde446 : SUCCESS in 2m 42s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/aa48e6f3501544dda9a3143aebb48a56 : SUCCESS in 38m 04s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e394723bd1b6467db396650782f01ff9 : SUCCESS in 37m 32s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2791bb8fb04c42f08abaf18116222311 : SUCCESS in 17m 38s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/a8ef1d4194f24976929945eb4ad53afa : FAILURE in 21m 46s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/eee6e0001a0b432b851d9a4907e28f19 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":2},{"id":"1aba83e8f488580f35780ab963ad76459fb8b189","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-04 11:20:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6100bce46db15140e6c791aae89c3111238069b","revisions":{"267a7607cfe53f8934cf3c3fb6efa93895837844":{"kind":"REWORK","_number":1,"created":"2026-08-28 13:51:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/1002760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/60/1002760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/60/1002760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/60/1002760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/60/1002760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/60/1002760/1"}}},"commit":{"parents":[{"commit":"04cc82cf2c71c735c3180c1c55e8e5224574ecd0","subject":"Unify template renderring for Agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04cc82cf2c71c735c3180c1c55e8e5224574ecd0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-28 13:51:09.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-28 13:51:09.000000000","tz":120},"subject":"Properly pass hostname and container name on pre-create","message":"Properly pass hostname and container name on pre-create\n\nWhen scheduler does pre-create a record in /backups it must supply\nrequired fields, like `hostname`, `container` and `backup_name`.\nWe were missing `hostname` and `container` name resolution could be\nincorrect in some cases. This was resulting in API failing with\nBad request during validation.\n\nChange-Id: I80eb5d1a63d6d2caf150665642871ec1268cab8f\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/267a7607cfe53f8934cf3c3fb6efa93895837844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/267a7607cfe53f8934cf3c3fb6efa93895837844"}]},"branch":"refs/heads/master"},"f6100bce46db15140e6c791aae89c3111238069b":{"kind":"REWORK","_number":2,"created":"2026-08-28 16:23:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/1002760/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/60/1002760/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/60/1002760/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/60/1002760/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/60/1002760/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/60/1002760/2"}}},"commit":{"parents":[{"commit":"04cc82cf2c71c735c3180c1c55e8e5224574ecd0","subject":"Unify template renderring for Agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04cc82cf2c71c735c3180c1c55e8e5224574ecd0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-28 13:51:09.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-28 16:23:06.000000000","tz":120},"subject":"Properly pass hostname and container name on pre-create","message":"Properly pass hostname and container name on pre-create\n\nWhen scheduler does pre-create a record in /backups it must supply\nrequired fields, like `hostname`, `container` and `backup_name`.\nWe were missing `hostname` and `container` name resolution could be\nincorrect in some cases. This was resulting in API failing with\nBad request during validation.\n\nChange-Id: I80eb5d1a63d6d2caf150665642871ec1268cab8f\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f6100bce46db15140e6c791aae89c3111238069b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f6100bce46db15140e6c791aae89c3111238069b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
