)]}'
{"id":"openstack%2Ffreezer~1002792","triplet_id":"openstack%2Ffreezer~master~Ieb736be91761e31af9a66112a91c100e2cd014f8","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Ieb736be91761e31af9a66112a91c100e2cd014f8","subject":"Add missing backup_id to agent config","status":"MERGED","created":"2026-08-28 16:33:18.000000000","updated":"2026-08-28 18:01:35.000000000","submitted":"2026-08-28 18:00:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002792","meta_rev_id":"aaeac40a361df93f12aea700e09909aef10c4227","_number":1002792,"virtual_id_number":1002792,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 18:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-08-28 16:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-08-28 16:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 17:15:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59e149875ef8fc6c81c6b7c768c4a4b9b37fa001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 16:33:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d5299c6f7370efda890403b0f19df567e912290","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-28 16:48:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6baa508e2e08210a09202c2d26e2d867bf1798f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 17:15:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb307bc2451b4f8da9e9195fe33b9d1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995a29a6e6024f44ba82937596606ac9 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c2d344b912a42428680498f67e88198 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d44993d70547b2aecfeb83eff5c62d : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4ae1c01e746463a9857bc1f07719bd3 : SUCCESS in 6m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7295327fa2249e3a57890c59a0b1b93 : SUCCESS in 7m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d556fad08564494a089cfdc574e71bd : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/df2487f3c1c843d5934cca6d13e3197e : SUCCESS in 4m 22s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e3aacb919fed45a9891b435f413f8db0 : SUCCESS in 36m 36s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/d0f7d44f5ae24125bfc2347793d7ec01 : SUCCESS in 36m 00s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ae6826adbcd47c1925f072d9e020d6d : SUCCESS in 15m 12s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/4f9210a267c9462687d38e8ddad72d70 : SUCCESS in 39m 33s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/2b6cf1ee5e3142e1bebac5c5a10a14c3 : SUCCESS in 23m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2f3424c993316a2345580b824a3c08af5f25cf45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 17:16:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78e2446a52dde2b7b96f28d364e24b1195d37997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:00:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5e89166cde4347b15dca0b9a0b853e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25096b2829c4e29a2a0705990ec162f : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e1832edcdf2481bbb7d3f8b202b4638 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5b5d23069a486b8d32d866da56945b : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d08cc2430df4f6588c59cd859790204 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8316ed595f084effbac48f18ca1ba596 : SUCCESS in 4m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73c7c2e9aa5e4f7095e8235535acba3c : SUCCESS in 4m 25s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9ed680ff4bdc4b8da0944bbb57150023 : SUCCESS in 14m 46s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/11d31b88b8a04d4a8598be939328fe0b : SUCCESS in 30m 58s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d95cbfb620b3441fb04de838b15e4808 : SUCCESS in 35m 08s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/582ae04b857f4125bf3b10aa7656f645 : SUCCESS in 19m 40s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/e4bb19b883cc4de6a5b5f83e34df2a75 : SUCCESS in 37m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d0a459f14865e495988c3782b815989c148f32dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:00:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aaeac40a361df93f12aea700e09909aef10c4227","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 18:01:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b10f74cd32b4c939cd436df77f0db69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2379d18e9cfc4a7d94d47f6d41fcdeab : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","revisions":{"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3":{"kind":"REWORK","_number":1,"created":"2026-08-28 16:33:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/1002792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/92/1002792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/92/1002792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/92/1002792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/92/1002792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/92/1002792/1"}}},"commit":{"parents":[{"commit":"c84243d8fc98faa137886b0daef1637d549b005a","subject":"Merge \"mongo: use a valid MongoDB URI for the client connection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c84243d8fc98faa137886b0daef1637d549b005a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-28 16:33:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-28 16:33:11.000000000","tz":120},"subject":"Add missing backup_id to agent config","message":"Add missing backup_id to agent config\n\nWe added support for pre-created backup record, which supposed to\nbe added into metadata of cindernative backups, however forgot\nto add respective argument to agent config.\n\nChange-Id: Ieb736be91761e31af9a66112a91c100e2cd014f8\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
