)]}'
{"id":"openstack%2Ffreezer~1002923","triplet_id":"openstack%2Ffreezer~master~I3e24f3de8337373289edc7c7a380aa08cf9f4065","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-31 16:41:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-31 18:04:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e24f3de8337373289edc7c7a380aa08cf9f4065","subject":"bindep: Remove stale markers","status":"MERGED","created":"2026-08-30 02:03:03.000000000","updated":"2026-08-31 18:05:55.000000000","submitted":"2026-08-31 18:04:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002923","meta_rev_id":"06fbad9cbef5a4dee30d9b3aed83ef635076433c","_number":1002923,"virtual_id_number":1002923,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 18:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-08-31 16:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-08-31 16:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 02:04:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-08-30 04:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0481cf0779ed29a2eca2960d2ef7c6357cfbb0b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:03:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b37515387a08b14a86709d5eff2afaf3978c927","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:03:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bf688e80d9333c319a3d95267d9b2e046f8a2d35","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:04:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"38b07537ddb7cb731c356adb8799535566dac375","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:04:33.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"525c1c17b6615dc5cf5bd3af3d61ca09b8682ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 04:01:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d266c20fb8e44cd97263dddf961e4ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7274f73b670942a6b9768b3337c6e84b : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5861213c73ee45deb708e5dcd31f10e3 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b616f9d70cd1476b9d7e958f9e51ba59 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70fb4cec5bc64b3daadcaeed31605c50 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a3b8e5f189a4abfbd73f83a18edefe2 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82e6a657bffe4488a008f384fffce860 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/199743fb7861487fae758ddb9a3245db : SUCCESS in 4m 07s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/c5f93fdb21fe488e8e9e9c9f17a24cdf : SUCCESS in 36m 16s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/688156a7ff8742e2bdcab9a7de6d0822 : SUCCESS in 41m 06s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36eeb9adfbb14dd8848ea11d626bd780 : FAILURE in 22m 12s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/6dd57d7a9ffd42cbb7a2a5d28b6cce3a : SUCCESS in 43m 42s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/021d25dc9c0a48c59d1fd46d058fb937 : FAILURE in 28m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6d3530eaf09a04decb71cb180e201bb2e5e16bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 12:00:38.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"e269a4bee1efb6ff6d068d1d62f57c6b12d70b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 12:44:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89cf215d5b26408ea2dbd1a932c98edf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790abfb910ab4ae0bf6142dc3ef8331b : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee8905bd4f30438abc28dc97b54fd8f4 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde6a65d868249f08a5bfb02c30cb807 : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ae8951e161c40aaa6c33a17d15f201a : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a38fad506ecc46a797755b0cf0ee28a5 : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c3ea38963d9492387143eab814d8035 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0691faffedf64ef88f1dba20d4d8ea63 : SUCCESS in 2m 49s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/06f4212bf34e485992ae7ed94599042c : SUCCESS in 14m 49s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/c654da4d8bdd4590a44b1c937b51545b : SUCCESS in 36m 48s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79cd48c3d8174c069662b3c4f9fa0464 : SUCCESS in 32m 42s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/27bff7ca05bd4ccda04cf1e5f542d61a : SUCCESS in 38m 49s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/49f98648db8f44a3a5141df8a47dadfe : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":4},{"id":"40365c456c018121b2ddf48d93bc28d51d95812c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 16:41:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"84de785746fee6f333128ef4b1a57649b730b332","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:41:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1007f58ebb487f4b4b1fd04e1b3b7b11280566db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:04:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4a1a083b258436293f3ded50d3d87ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef54547562b40839c7f3c29717e037a : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c655ed4e9be94df6b839265beabdd736 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/207f831a30b34a2488314b5a65f5d919 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a9798ce17a74bbab26b5a0ffa45249d : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86d27691a7ec4c7699b5114892f52735 : SUCCESS in 5m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/837e25e0f0d04c9f9e0f0c0d88969087 : SUCCESS in 2m 37s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/80081e5b01bb43b88d4ce3ac03f2ed5d : SUCCESS in 21m 47s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/dc52ed61d1d946579e3b22ba46357e9f : SUCCESS in 23m 28s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8000d36fcc743e9bd1c42587a6a95b4 : SUCCESS in 21m 31s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/e1ac11a17a7c49d4a8ba29a607343fe4 : SUCCESS in 26m 29s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/911cc09bffb0402bba39efb8cd100fa0 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":4},{"id":"8cff114079f7a21ab288ff8edfcf38c37bd43aeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:04:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"06fbad9cbef5a4dee30d9b3aed83ef635076433c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:05:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f425199e13c6433a9c3bb38392bb62b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328754e72b3f49c1b9681173380b05de : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2fc7e22a8544566934de19391e75adb : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"38c360ff5c4c3f109d6db35a8fb014e2aa67b6c8","revisions":{"be75d90bd47d7ad5318d0634e25b51faf9a7026c":{"kind":"REWORK","_number":1,"created":"2026-08-30 02:03:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/23/1002923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/1002923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/1002923/1"}}},"commit":{"parents":[{"commit":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","subject":"Add missing backup_id to agent config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:03:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:03:03.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I3e24f3de8337373289edc7c7a380aa08cf9f4065\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/be75d90bd47d7ad5318d0634e25b51faf9a7026c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/be75d90bd47d7ad5318d0634e25b51faf9a7026c"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"14cbb0052b51cff6c44354a0c336fb846e852b24":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-30 02:03:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/23/1002923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/1002923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/1002923/2"}}},"commit":{"parents":[{"commit":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","subject":"Add missing backup_id to agent config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:03:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:03:31.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nChange-Id: I3e24f3de8337373289edc7c7a380aa08cf9f4065\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/14cbb0052b51cff6c44354a0c336fb846e852b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/14cbb0052b51cff6c44354a0c336fb846e852b24"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"98463897b7a7c208744aeedc43849dd84bad14ea":{"kind":"REWORK","_number":3,"created":"2026-08-30 02:04:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/23/1002923/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/1002923/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/1002923/3"}}},"commit":{"parents":[{"commit":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","subject":"Add missing backup_id to agent config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:03:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:04:12.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nChange-Id: I3e24f3de8337373289edc7c7a380aa08cf9f4065\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/98463897b7a7c208744aeedc43849dd84bad14ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/98463897b7a7c208744aeedc43849dd84bad14ea"}]},"branch":"refs/heads/master"},"38c360ff5c4c3f109d6db35a8fb014e2aa67b6c8":{"kind":"REWORK","_number":4,"created":"2026-08-30 12:00:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/23/1002923/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/23/1002923/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/23/1002923/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/23/1002923/4"}}},"commit":{"parents":[{"commit":"cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3","subject":"Add missing backup_id to agent config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cd070e8d7ea1ac7d6de887ea0374c9bc56c28de3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:03:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 12:00:38.000000000","tz":0},"subject":"bindep: Remove stale markers","message":"bindep: Remove stale markers\n\nRemove markers for old operating system versions which are no longer\nsupported.\n\nChange-Id: I3e24f3de8337373289edc7c7a380aa08cf9f4065\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/38c360ff5c4c3f109d6db35a8fb014e2aa67b6c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/38c360ff5c4c3f109d6db35a8fb014e2aa67b6c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
