)]}'
{"id":"openstack%2Ffreezer~1003929","triplet_id":"openstack%2Ffreezer~master~I724a144fbcc17ae8ee03a65bdfe36404ec632c1a","project":"openstack/freezer","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-04 11:58:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I724a144fbcc17ae8ee03a65bdfe36404ec632c1a","subject":"Improve datetime parsing compatability to ISO 8601","status":"MERGED","created":"2026-09-04 09:32:57.000000000","updated":"2026-09-04 12:48:03.000000000","submitted":"2026-09-04 12:46:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003929","meta_rev_id":"15d2847219d601a3c3b0131043c5209bffb26a5c","_number":1003929,"virtual_id_number":1003929,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 12:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"all":[{"value":1,"date":"2026-09-04 11:56:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":2,"date":"2026-09-04 11:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},{"value":1,"date":"2026-09-04 11:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 10:20:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 11:56:57.000000000","updated_by":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"REVIEWER"}],"messages":[{"id":"c98f1c5bb65c338f679ab84ca036f554ad2102d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-04 09:32:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ace7229d3ddaccc5e93618c3c5c4ef28bb32cd52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 10:20:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd58480f88a847d6b5bb0542d4b0ee8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1ce657e56941379e366f04d3c0e46b : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/481403905ca54bfea56621d577c97003 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e47598545cb442a98880239db4be3e1 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/751352e6db6d4c3f92331290db71df39 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2379e4ee74d94a5fae7c2ba6777250d7 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2423f457b8b84461aad7acff8c462373 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/560e2c2acf6844ea83f572b252284557 : SUCCESS in 4m 04s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/5e6bdb06f8884a03a7788f4cfce4b80c : SUCCESS in 32m 48s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/72a46215f559477e836473effb4ca7c8 : SUCCESS in 42m 11s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/795084a41a22406bbd12d09e488a8a78 : SUCCESS in 35m 36s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/9b410b195f604abd9d1224a497ccbd60 : SUCCESS in 42m 13s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/5681a2e113e448ecaf90852083c9ee01 : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6cf4e62e71b337bf4b1efce1ad85b4ed56e6bd6c","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-09-04 11:56:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6af2eb679476cf808daa4e68bada003d187c0d42","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-04 11:58:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1fcab776e544f067fd7c6ead4c67bc8d7939a8e5","author":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"date":"2026-09-04 11:59:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc1188d0907b8dafc9966fd227f9cb6817a9d4ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:59:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"021c595525c6077ae3e0266d0ddadfe73c945c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 12:46:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f2a64e4021149038c70cc56dfc43775\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f8369bffde4d92b82bd2e90d841cd0 : SUCCESS in 2m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb01e9a54084cf3adcab84923af5afb : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa190b8388e49debdf09bef93f43cc7 : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/020663d6f323461e94f66b296e0ace3c : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61694699c133467ca0f183af2ad8c9aa : SUCCESS in 4m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/43002d4dce7e40ae9ee1b1680555a60d : SUCCESS in 3m 03s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/5f4d7362ed824a0ea884743510eac43f : SUCCESS in 35m 05s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/8b81f935888c4e58842ba943a415131c : SUCCESS in 32m 41s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9aefee4c80c4264a7062bfa0a01bf2c : SUCCESS in 37m 28s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/2cd9819b441147ebbc318438be8f5abe : SUCCESS in 42m 35s\n- freezer-centos-10-stream https://zuul.opendev.org/t/openstack/build/068903fdda8b4aa08a4d3dd117bed42f : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"45124a2a00a6820ef31f82733523dd99d6efda7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 12:46:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"15d2847219d601a3c3b0131043c5209bffb26a5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 12:48:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb230cb307745eabb20c58d3e109b47\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4123860f55ad4f4c982e13bb68281ec5 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed2edc10cb9102b8d5b15baed97bda1d12ac9d47","revisions":{"ed2edc10cb9102b8d5b15baed97bda1d12ac9d47":{"kind":"REWORK","_number":1,"created":"2026-09-04 09:32:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/29/1003929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/29/1003929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/29/1003929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/29/1003929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/29/1003929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/29/1003929/1"}}},"commit":{"parents":[{"commit":"410eb878e14e028a7af7ccf6521e1262ff1fa09f","subject":"Merge \"Skip storage initialization for cindernative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/410eb878e14e028a7af7ccf6521e1262ff1fa09f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-04 09:32:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-04 09:32:49.000000000","tz":120},"subject":"Improve datetime parsing compatability to ISO 8601","message":"Improve datetime parsing compatability to ISO 8601\n\nWith [1] we landed a regression in parsing datetime in utils, when\ntotally valid datetime\naccording to ISO 8601 would raise ValueError.\n\nThis patch makes freezer to leverage oslo_utils.timeutils\nfor timedate input parsing, which allows us to accept TZ indicators\nas well as fractional timestamps.\n\n[1] https://review.opendev.org/c/openstack/freezer/+/1000670/16/freezer/utils/utils.py\n\nChange-Id: I724a144fbcc17ae8ee03a65bdfe36404ec632c1a\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ed2edc10cb9102b8d5b15baed97bda1d12ac9d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ed2edc10cb9102b8d5b15baed97bda1d12ac9d47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
