)]}'
{"id":"openstack%2Ffreezer~456758","triplet_id":"openstack%2Ffreezer~master~I2d43f14943fc52fd6b6cb994251c97e00d3eb863","project":"openstack/freezer","branch":"master","topic":"bug/1682606","hashtags":[],"change_id":"I2d43f14943fc52fd6b6cb994251c97e00d3eb863","subject":"Force Swift storage driver to always get a new client","status":"MERGED","created":"2017-04-13 20:55:25.000000000","updated":"2017-04-14 13:03:07.000000000","submitted":"2017-04-14 11:06:01.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a6eaba884627c0b14a382093abf2c6a8fa68fc1","_number":456758,"virtual_id_number":456758,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-04-14 11:06:01.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":0,"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},{"value":0,"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},{"value":0,"_account_id":14509,"name":"Pierre Mathieu","email":"pierre-arthur.mathieu@hpe.com","username":"slashme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-04-14 10:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":0,"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},{"value":1,"date":"2017-04-13 21:10:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},{"value":0,"_account_id":14509,"name":"Pierre Mathieu","email":"pierre-arthur.mathieu@hpe.com","username":"slashme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":1,"date":"2017-04-14 10:44:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},{"value":0,"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},{"value":0,"date":"2017-04-14 10:25:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14509,"name":"Pierre Mathieu","email":"pierre-arthur.mathieu@hpe.com","username":"slashme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},{"_account_id":14509,"name":"Pierre Mathieu","email":"pierre-arthur.mathieu@hpe.com","username":"slashme"},{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-13 21:10:20.000000000","updated_by":{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},"reviewer":{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},"state":"REVIEWER"},{"updated":"2017-04-14 10:25:30.000000000","updated_by":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"},{"updated":"2017-04-14 10:25:39.000000000","updated_by":{"_account_id":14509,"name":"Pierre Mathieu","email":"pierre-arthur.mathieu@hpe.com","username":"slashme"},"reviewer":{"_account_id":14509,"name":"Pierre Mathieu","email":"pierre-arthur.mathieu@hpe.com","username":"slashme"},"state":"REVIEWER"},{"updated":"2017-04-14 10:44:20.000000000","updated_by":{"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},"reviewer":{"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},"state":"REVIEWER"},{"updated":"2017-04-14 11:06:01.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ea66fa3855e91cf2bbe4a23f8496016cedb2493d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-04-13 20:55:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcd9b5038579cc544c104894d081e2429b200920","author":{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},"date":"2017-04-13 21:10:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ddc50492d31d91865e868eaa2230908b8dfc01f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-13 21:18:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-freezer-docs-ubuntu-xenial http://docs-draft.openstack.org/58/456758/1/check/gate-freezer-docs-ubuntu-xenial/06f13c0//doc/build/html/ : SUCCESS in 1m 23s\n- gate-freezer-pep8-ubuntu-xenial http://logs.openstack.org/58/456758/1/check/gate-freezer-pep8-ubuntu-xenial/59e7035/ : SUCCESS in 2m 11s\n- gate-freezer-python27-ubuntu-xenial http://logs.openstack.org/58/456758/1/check/gate-freezer-python27-ubuntu-xenial/5dc6cb7/ : SUCCESS in 2m 24s\n- gate-freezer-python35 http://logs.openstack.org/58/456758/1/check/gate-freezer-python35/196a63f/ : SUCCESS in 1m 20s\n- gate-osbackup-freezer-dsvm-ubuntu-xenial http://logs.openstack.org/58/456758/1/check/gate-osbackup-freezer-dsvm-ubuntu-xenial/2ec9b83/ : SUCCESS in 22m 15s\n- gate-freezer-pylint-ubuntu-xenial http://logs.openstack.org/58/456758/1/check/gate-freezer-pylint-ubuntu-xenial/91d0493/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d3f5675d93620a4b1e1d0c1e394f56a9d6e30ccc","author":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"date":"2017-04-14 10:25:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ee12e7b8b32b38510c73cab6af2a24af0468532","author":{"_account_id":14340,"name":"Memo Garcia","email":"guillermo.ramirez-garcia@hp.com","username":"memogarcia"},"date":"2017-04-14 10:44:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6331685fddc804056b51af2b05104cf5dfc4f67e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-14 10:44:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"d032eb5b36c658ae68d7124a5119f70df0c3d5cb","date":"2017-04-14 11:06:01.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"e13a1286e2874da38368ea8b3d42f0716e341a81","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-14 11:06:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-freezer-docs-ubuntu-xenial http://docs-draft.openstack.org/58/456758/1/gate/gate-freezer-docs-ubuntu-xenial/59d645b//doc/build/html/ : SUCCESS in 2m 38s\n- gate-freezer-pep8-ubuntu-xenial http://logs.openstack.org/58/456758/1/gate/gate-freezer-pep8-ubuntu-xenial/3ac4bdf/ : SUCCESS in 2m 18s\n- gate-freezer-python27-ubuntu-xenial http://logs.openstack.org/58/456758/1/gate/gate-freezer-python27-ubuntu-xenial/4328455/ : SUCCESS in 1m 13s\n- gate-freezer-python35 http://logs.openstack.org/58/456758/1/gate/gate-freezer-python35/57cd08b/ : SUCCESS in 2m 25s\n- gate-osbackup-freezer-dsvm-ubuntu-xenial http://logs.openstack.org/58/456758/1/gate/gate-osbackup-freezer-dsvm-ubuntu-xenial/4b6e0c5/ : SUCCESS in 21m 15s\n- gate-freezer-pylint-ubuntu-xenial http://logs.openstack.org/58/456758/1/gate/gate-freezer-pylint-ubuntu-xenial/81b795b/ : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"135ef777583124e02098b1b6045722060e8e60ad","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-04-14 13:03:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ocata as commit 6b319c28a8e42e60ce22f42718134532d03d4af3","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c83a3f7c3ce699b377210fed8e54c73ddf18b96","revisions":{"4c83a3f7c3ce699b377210fed8e54c73ddf18b96":{"kind":"REWORK","_number":1,"created":"2017-04-13 20:55:25.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/58/456758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/58/456758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/58/456758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/58/456758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/58/456758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/58/456758/1"}}},"commit":{"parents":[{"commit":"bcfe5e2f5504db6964eb1e559e3f4b7ceeb09649","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bcfe5e2f5504db6964eb1e559e3f4b7ceeb09649"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2017-04-13 20:55:04.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2017-04-13 20:55:24.000000000","tz":-240},"subject":"Force Swift storage driver to always get a new client","message":"Force Swift storage driver to always get a new client\n\nAt the moment, the Swift storage codebase does not have a consistent\nway of using the Swift client, as it switches from grabbing an\nexisting client and creating new ones.\n\nDue to the architecture of how restores happen, there are several\ntimes where the same storage instance is held by the main process,\nthe process which reads from the storage and the engine (to retrieve\nmetadata).  All of these accesses result in unexpected behaviour which\ncan raise SSL errors such as: DECRYPTION_FAILED_OR_BAD_RECORD_MAC\n\nThe overall cost of creating a new client for every request is very\nminimal and it eliminates all possibility of multithreading issues\nthat occur as per suggested by the author of the requests Python\nlibrary[0], which is used by the Swift client.\n\nThis patch eliminates all calls to create_swift() indirectly and makes\nsure that all accesses to the Swift client create a new instance. In\nmy local testing, there was no noticable performance change but\nno more SSL failures.\n\nCloses-Bug: #1682606\n\n[0]: https://github.com/kennethreitz/requests/issues/1906#issuecomment-201238170\n\nChange-Id: I2d43f14943fc52fd6b6cb994251c97e00d3eb863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/4c83a3f7c3ce699b377210fed8e54c73ddf18b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/4c83a3f7c3ce699b377210fed8e54c73ddf18b96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
