)]}'
{"id":"openstack%2Ffreezer~526667","triplet_id":"openstack%2Ffreezer~master~I4625d55a768f1ad0762fc2d8554998825f0d6716","project":"openstack/freezer","branch":"master","topic":"goal-split-tempest-plugins","hashtags":[],"change_id":"I4625d55a768f1ad0762fc2d8554998825f0d6716","subject":"Remove bundled intree freezer tempest plugin","status":"MERGED","created":"2017-12-08 12:33:54.000000000","updated":"2017-12-10 22:32:47.000000000","submitted":"2017-12-10 22:32:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"526667-1512945167969-41e1d265","meta_rev_id":"fb122b934f8582ed2f013013c54b26e428177b89","_number":526667,"virtual_id_number":526667,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":0,"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},{"value":2,"date":"2017-12-10 22:32:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},"all":[{"value":0,"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":2,"date":"2017-12-10 00:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"all":[{"value":1,"date":"2017-12-10 22:25:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"value":0,"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-10 00:49:31.000000000","updated_by":{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},"reviewer":{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},"state":"REVIEWER"},{"updated":"2017-12-10 22:25:54.000000000","updated_by":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"},{"updated":"2017-12-10 22:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7d22f30d9113e7a797ccd08ad771c9e7405dfe6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-12-08 12:33:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9519977f2d3923aa95b323d994016e5d4205174e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-08 14:24:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/67/526667/1/check/openstack-tox-pep8/4e3e87d/ : FAILURE in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/67/526667/1/check/openstack-tox-py27/e257241/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/67/526667/1/check/openstack-tox-py35/9e9f55b/ : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/67/526667/1/check/build-openstack-sphinx-docs/68a5329/html/ : SUCCESS in 5m 57s\n- openstack-tox-pylint http://logs.openstack.org/67/526667/1/check/openstack-tox-pylint/ef415a3/ : SUCCESS in 3m 56s\n- freezer-ubuntu-xenial http://logs.openstack.org/67/526667/1/check/freezer-ubuntu-xenial/0c915ad/ : SUCCESS in 28m 35s (non-voting)\n- freezer-centos-7 http://logs.openstack.org/67/526667/1/check/freezer-centos-7/090c606/ : FAILURE in 6m 26s (non-voting)\n- freezer-opensuse-423 http://logs.openstack.org/67/526667/1/check/freezer-opensuse-423/8b376cb/ : FAILURE in 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5d50bba5198c2facf27595c50c722889f684b7c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-12-08 16:45:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d0357977b64498fa5d8f429c9b9a9940f2e26d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-08 18:22:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/67/526667/2/check/openstack-tox-pep8/ce087b0/ : SUCCESS in 7m 00s\n- openstack-tox-py27 http://logs.openstack.org/67/526667/2/check/openstack-tox-py27/b0a9e3c/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/67/526667/2/check/openstack-tox-py35/efb692d/ : SUCCESS in 3m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/67/526667/2/check/build-openstack-sphinx-docs/9ffecdd/html/ : SUCCESS in 3m 33s\n- openstack-tox-pylint http://logs.openstack.org/67/526667/2/check/openstack-tox-pylint/f38991f/ : SUCCESS in 3m 22s\n- freezer-ubuntu-xenial http://logs.openstack.org/67/526667/2/check/freezer-ubuntu-xenial/dcb338f/ : SUCCESS in 29m 37s (non-voting)\n- freezer-centos-7 http://logs.openstack.org/67/526667/2/check/freezer-centos-7/4096154/ : FAILURE in 10m 23s (non-voting)\n- freezer-opensuse-423 http://logs.openstack.org/67/526667/2/check/freezer-opensuse-423/5a12cfb/ : FAILURE in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cca5421b8609bf17906809be083b071beeadf6ce","author":{"_account_id":22405,"name":"Vitaliy Nogin","email":"shadoff1982@gmail.com","username":"vnogin"},"date":"2017-12-10 00:49:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4333dcae1c00556c9b8cb498ad9d32d85910132d","author":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"date":"2017-12-10 22:25:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"23283fb12c2be2dc36c5589491d8ac1ab3f53179","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-10 22:26:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60189c93c85a604b3d54173ca0ab918007eb3e27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-10 22:32:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/67/526667/2/gate/openstack-tox-pep8/cc1f67f/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/67/526667/2/gate/openstack-tox-py27/e76ac4d/ : SUCCESS in 5m 47s\n- openstack-tox-py35 http://logs.openstack.org/67/526667/2/gate/openstack-tox-py35/47f6aac/ : SUCCESS in 3m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/67/526667/2/gate/build-openstack-sphinx-docs/10fa1fa/html/ : SUCCESS in 3m 30s\n- openstack-tox-pylint http://logs.openstack.org/67/526667/2/gate/openstack-tox-pylint/e99488c/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1b6d28c8562eca77b8184d03f382ccbb6a56e399","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-10 22:32:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"01da61b237032b0f21a9b3583a69bedd8b7a2417","revisions":{"78da85a429ab286ae67e9493b4d746660f9c9bca":{"kind":"REWORK","_number":1,"created":"2017-12-08 12:33:54.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/67/526667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/67/526667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/67/526667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/67/526667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/67/526667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/67/526667/1"}}},"commit":{"parents":[{"commit":"3bbebaa4161f9f3c7905feee6cde561153875b25","subject":"Fix gate jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3bbebaa4161f9f3c7905feee6cde561153875b25"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-12-08 12:29:24.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-12-08 12:33:45.000000000","tz":330},"subject":"Remove bundled intree freezer tempest plugin","message":"Remove bundled intree freezer tempest plugin\n\n* https://review.openstack.org/#/c/526664/ moves the intree tempest\n  plugin to freezer-tempest-plugin repo.\n\n* Excluding freezer/tests/freezer_tempest_plugin/tests/api/test_version.py as\n  it is dependent on freezer project and moving under integration tests.\n\nDepends-On: I6967f915758728827e8ddcd1a45a7023904b694e\nChange-Id: I4625d55a768f1ad0762fc2d8554998825f0d6716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/78da85a429ab286ae67e9493b4d746660f9c9bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/78da85a429ab286ae67e9493b4d746660f9c9bca"}]},"branch":"refs/heads/master"},"01da61b237032b0f21a9b3583a69bedd8b7a2417":{"kind":"REWORK","_number":2,"created":"2017-12-08 16:45:15.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/67/526667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/67/526667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/67/526667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/67/526667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/67/526667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/67/526667/2"}}},"commit":{"parents":[{"commit":"3bbebaa4161f9f3c7905feee6cde561153875b25","subject":"Fix gate jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3bbebaa4161f9f3c7905feee6cde561153875b25"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-12-08 12:29:24.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-12-08 16:45:01.000000000","tz":330},"subject":"Remove bundled intree freezer tempest plugin","message":"Remove bundled intree freezer tempest plugin\n\n* https://review.openstack.org/#/c/526664/ moves the intree tempest\n  plugin to freezer-tempest-plugin repo.\n\n* Excluding freezer/tests/freezer_tempest_plugin/tests/api/test_version.py as\n  it is dependent on freezer project and moving under integration tests.\n\nDepends-On: I6967f915758728827e8ddcd1a45a7023904b694e\nChange-Id: I4625d55a768f1ad0762fc2d8554998825f0d6716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/01da61b237032b0f21a9b3583a69bedd8b7a2417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/01da61b237032b0f21a9b3583a69bedd8b7a2417"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
