)]}'
{"id":"openstack%2Ffreezer~667091","triplet_id":"openstack%2Ffreezer~master~I37c8620e53f45c3bfce2c018f65149c8505cb956","project":"openstack/freezer","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I37c8620e53f45c3bfce2c018f65149c8505cb956","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-06-24 13:21:15.000000000","updated":"2019-07-15 01:44:38.000000000","submitted":"2019-07-15 01:44:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667091-1563155078801-4e3e88c1","meta_rev_id":"a183782c315fc89def73bfea3882abaab8e32674","_number":667091,"virtual_id_number":667091,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-07-15 01:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},"all":[{"value":2,"date":"2019-07-15 01:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},{"value":2,"date":"2019-07-15 01:04:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"date":"2019-06-25 13:35:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":0,"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},{"value":1,"date":"2019-07-15 01:04:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 13:35:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-07-15 01:00:48.000000000","updated_by":{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},"reviewer":{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},"state":"REVIEWER"},{"updated":"2019-07-15 01:04:32.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2019-07-15 01:44:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3dab48937485148e23fb7ffa9db2fa420caa923e","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-24 13:21:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c81f54006e899de4a781ff5f2974c02b6d78e259","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/91/667091/1/check/openstack-tox-lower-constraints/fce907d/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/91/667091/1/check/openstack-tox-pep8/d0806af/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/91/667091/1/check/openstack-tox-py27/89cd674/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/91/667091/1/check/openstack-tox-py36/e56bd07/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/91/667091/1/check/openstack-tox-py37/4ca74be/ : SUCCESS in 6m 27s\n- openstack-tox-docs http://logs.openstack.org/91/667091/1/check/openstack-tox-docs/86bf318/html/ : SUCCESS in 3m 53s\n- build-openstack-releasenotes http://logs.openstack.org/91/667091/1/check/build-openstack-releasenotes/ede7576/html/ : SUCCESS in 3m 34s\n- freezer-ubuntu http://logs.openstack.org/91/667091/1/check/freezer-ubuntu/f415ac8/ : SUCCESS in 38m 46s\n- freezer-centos-7 http://logs.openstack.org/91/667091/1/check/freezer-centos-7/de377aa/ : SUCCESS in 50m 08s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/91/667091/1/check/openstack-tox-pylint/d76fefb/ : SUCCESS in 4m 50s\n- freezer-tempest-agent http://logs.openstack.org/91/667091/1/check/freezer-tempest-agent/156dda1/ : SUCCESS in 33m 58s\n- freezer-tempest-scheduler http://logs.openstack.org/91/667091/1/check/freezer-tempest-scheduler/71f8454/ : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"73c75a54f08042bd7958f2fb9a89beaa3f60e5c6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-25 11:45:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e19c7ed07ca2a9eea9064743ea403ba7c56529b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-25 13:35:44.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75bb44f94164de656988f444e98ff04c300168d3","author":{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},"date":"2019-07-15 01:00:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"921735ce7cfb68d5b863564f626ce3f084e65f54","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2019-07-15 01:04:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8116cfc800912a80daf9ac25f34f11551a8f2224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 01:04:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20bf7b65bcb17dea0b3b231c3b84236d8244c6c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 01:44:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"699be786edf4feb287cec31510c0fc938e8f141f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 01:44:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/91/667091/1/gate/openstack-tox-lower-constraints/1b39b81/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/91/667091/1/gate/openstack-tox-pep8/12f19e2/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/91/667091/1/gate/openstack-tox-py27/cc43e5b/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/91/667091/1/gate/openstack-tox-py36/f576de5/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/91/667091/1/gate/openstack-tox-py37/9b21120/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/91/667091/1/gate/openstack-tox-docs/747b663/html/ : SUCCESS in 4m 07s\n- build-openstack-releasenotes http://logs.openstack.org/91/667091/1/gate/build-openstack-releasenotes/3353342/html/ : SUCCESS in 3m 44s\n- freezer-ubuntu http://logs.openstack.org/91/667091/1/gate/freezer-ubuntu/3f60e85/ : SUCCESS in 37m 30s\n- openstack-tox-pylint http://logs.openstack.org/91/667091/1/gate/openstack-tox-pylint/96adb07/ : SUCCESS in 4m 52s\n- freezer-tempest-agent http://logs.openstack.org/91/667091/1/gate/freezer-tempest-agent/aad4623/ : SUCCESS in 36m 09s\n- freezer-tempest-scheduler http://logs.openstack.org/91/667091/1/gate/freezer-tempest-scheduler/bc6c9b0/ : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79b064a079cd6daaa1c104d775c1c0e9197776a6","revisions":{"79b064a079cd6daaa1c104d775c1c0e9197776a6":{"kind":"REWORK","_number":1,"created":"2019-06-24 13:21:15.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/91/667091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/91/667091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/91/667091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/91/667091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/91/667091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/91/667091/1"}}},"commit":{"parents":[{"commit":"720232769b12bac126f4c066d10f41691c45568c","subject":"Merge \"Make the low-constraints job work and modify coverage version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/720232769b12bac126f4c066d10f41691c45568c"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 13:20:12.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 13:21:14.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I37c8620e53f45c3bfce2c018f65149c8505cb956\nStory: #2005924\nTask: #34207\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/79b064a079cd6daaa1c104d775c1c0e9197776a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/79b064a079cd6daaa1c104d775c1c0e9197776a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
