)]}'
{"id":"openstack%2Ffreezer~682744","triplet_id":"openstack%2Ffreezer~master~If9e8913e49d9ecf561a7e1824245a995fc6ac6db","project":"openstack/freezer","branch":"master","topic":"ipv6-only-deployment-and-testing","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-11-24 08:26:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-27 17:48:18.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-11-24 08:34:34.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"If9e8913e49d9ecf561a7e1824245a995fc6ac6db","subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","status":"MERGED","created":"2019-09-17 20:29:26.000000000","updated":"2025-11-27 17:50:35.000000000","submitted":"2025-11-27 17:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682744-ipv6-only-deployment-and-testing","meta_rev_id":"1ef26eaa256e7c5b2655835a57f8d7d9c9d2491d","_number":682744,"virtual_id_number":682744,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-27 17:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-11-24 08:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-23 12:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-11-24 08:26:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 01:38:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-23 12:24:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-11-23 12:24:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"967bd0f0075335cff2afa256006af00997b3de40","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 20:29:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"893bd4add02ec7e420709a49ec1aa82193c54990","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 20:31:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"94b0f0e18717220fa772dfc8bf40397de462067f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-17 20:31:29.000000000","message":"Topic set to ipv6-only-deployment-and-testing","accounts_in_message":[],"_revision_number":2},{"id":"36faef6d711d0a85f23d1a3657e55aadd582a0b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:35:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbb21e2244f0471cb88b4ebcee0a94ca : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0680a1ee50b4352859c7c33efd227ce : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/caba075ba5e6472f89f1b2a1ac233500 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9cb6df7438048b28aee09457a1b26eb : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f556c9cfc6944d78bd7ca8cbe5c49fc3 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9c7fa6c27c4353a596f3183a92d005 : SUCCESS in 6m 54s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/d6ea2711a01249819f4434f403a42306 : SUCCESS in 42m 46s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/1e11697f6dbe4d818bd3cb493daa51b2 : SUCCESS in 59m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6fb91ca9c364cd287390e3235f52909 : SUCCESS in 4m 07s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/458f33cb1c51471da6cbe5928e13cd49 : SUCCESS in 42m 53s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/cc2acfcfbf3747cea0a1e70f068224ca : SUCCESS in 34m 41s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bad775a3b1464ae3b5722ac3b2b38055 : FAILURE in 37m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b27cb331340c25436ab943eba9dc7dd6860e35c3","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-11-15 01:37:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2e0cb91e91377e800ef2dfa2217ce2acf00749dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 01:38:21.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"af3767b12f0cc0f858308dffb9ef391f43eea8f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-23 11:18:12.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1a6c07aaff66232ebbcd2ad4d32480b8706abe54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-23 11:18:37.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ffb95d36b7ba86bcba72d570085a5ee235191b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 11:18:44.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/freezer for 682744,4","accounts_in_message":[],"_revision_number":4},{"id":"75f95ecae3eec0fdbb1ee5d29c460414a35eb39b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-23 11:18:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7d9721431c3b0953db33c4282c1f17fea70a6acf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-23 11:19:34.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"e065420343b128189408ac8ebbc4ab54960d530a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 12:03:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13b8929a48e4d93bc9a2b35b1001e32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f0421e85ea42338d041a5bed969900 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cafad91cfa304ced8ad6401d0f41b1ec : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74e6f8a1c5854386a1a49d3cf885afd0 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c1f9293f52342b38cda9b8c93b195b8 : SUCCESS in 6m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/080fe7f48585494dac8e2b8a4b429024 : SUCCESS in 4m 00s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a7efd2c5cd6949b09511bacc17f0ff27 : SUCCESS in 33m 39s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/7267764632f348fc9410c33e1a1d1d36 : SUCCESS in 21m 34s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b95f31a90b74212a36121f61b3aa80a : SUCCESS in 21m 02s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b6f1a8129b554a469e406d0f30d90a42 : SUCCESS in 37m 21s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/0609f5e179c34e5ebd078d77790ccaae : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":7},{"id":"6b8e38240794de5e77accc48161aa04927421da5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-23 12:24:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"72dc3cf94b1642e5c18be545be762e37d8f808f0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-24 08:26:58.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cf62e5c80c7517d4cdffa9f0e6e83be73b290d92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:05:29.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9d05d26c0e33e58ca58d907ab14c05d9a660404b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:48:18.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f9e487643dc4a5a88e251a72c7fee9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf41e9e7f3ca4f42827f8c2a8bcd07d6 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3fc7838217439eb88d0c9c079ad71b : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2e4780a16f2466cba10a6ab99b97928 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7199218d8364d4d84371623a4b53ef6 : SUCCESS in 5m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd47a7497009425a910b29d985e03313 : SUCCESS in 2m 36s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/4a4caab1215a4caf8b5a5fdbaa2f883a : SUCCESS in 27m 46s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e8cb72a4eca446d6a51c9921a05e3d90 : SUCCESS in 35m 10s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adf6370ed2b941de84eba10b15b180a6 : SUCCESS in 15m 26s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/d52d73c6446c4c5aab9786f14492f931 : SUCCESS in 20m 33s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/04ea7c22af0d409c9f5b79910bc3c397 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":7},{"id":"acf977d0d3a4aa233c4cf235f4f19447bfc8d9ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:48:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1ef26eaa256e7c5b2655835a57f8d7d9c9d2491d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:50:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4311b9cec4b47878307f7b1bbd1f760\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ebd6fc30d3b41ada197ed96e8c4010a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b294b30892533c16c1a3235c801b18642c86f60e","revisions":{"971ec0787d27e6e3d8c83fb89d18e4462b906b49":{"kind":"REWORK","_number":1,"created":"2019-09-17 20:29:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/682744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/44/682744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/44/682744/1"}}},"commit":{"parents":[{"commit":"67abf372aa4fa8fa27a9091b10cbc3a6f2d1cebb","subject":"Update api-ref location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/67abf372aa4fa8fa27a9091b10cbc3a6f2d1cebb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:27:40.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:29:23.000000000","tz":0},"subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","message":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027freezer-tempest-ipv6-only\u0027 job in cyborg gate.\n\nDepends-On: https://review.opendev.org/#/c/682742\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: If9e8913e49d9ecf561a7e1824245a995fc6ac6db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/971ec0787d27e6e3d8c83fb89d18e4462b906b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/971ec0787d27e6e3d8c83fb89d18e4462b906b49"}]},"branch":"refs/heads/master"},"fa908a5237bce7a47fa95f4cbd0e072a191e6d91":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-17 20:31:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/682744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/44/682744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/44/682744/2"}}},"commit":{"parents":[{"commit":"67abf372aa4fa8fa27a9091b10cbc3a6f2d1cebb","subject":"Update api-ref location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/67abf372aa4fa8fa27a9091b10cbc3a6f2d1cebb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:27:40.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:31:04.000000000","tz":0},"subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","message":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027freezer-tempest-ipv6-only\u0027 job in freezer gate.\n\nDepends-On: https://review.opendev.org/#/c/682742\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: If9e8913e49d9ecf561a7e1824245a995fc6ac6db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fa908a5237bce7a47fa95f4cbd0e072a191e6d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fa908a5237bce7a47fa95f4cbd0e072a191e6d91"}]},"branch":"refs/heads/master"},"c3c8b1c76a5e5f41b8c1e9483baec6864a2a1520":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-15 01:37:05.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/44/682744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/44/682744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/44/682744/3"}}},"commit":{"parents":[{"commit":"e5a4d0cf75281bee076172dba81c27e09ec9bc38","subject":"Upgrade pylint to a new version that works with python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e5a4d0cf75281bee076172dba81c27e09ec9bc38"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:27:40.000000000","tz":0},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-11-15 01:37:05.000000000","tz":0},"subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","message":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027freezer-tempest-ipv6-only\u0027 job in freezer gate.\n\nDepends-On: https://review.opendev.org/#/c/682742\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: If9e8913e49d9ecf561a7e1824245a995fc6ac6db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c3c8b1c76a5e5f41b8c1e9483baec6864a2a1520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c3c8b1c76a5e5f41b8c1e9483baec6864a2a1520"}]},"branch":"refs/heads/master"},"a479b0421e12017d2f70b91e11061a7d5dcdb58d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-11-23 11:18:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/682744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/44/682744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/44/682744/4"}}},"commit":{"parents":[{"commit":"e5a4d0cf75281bee076172dba81c27e09ec9bc38","subject":"Upgrade pylint to a new version that works with python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e5a4d0cf75281bee076172dba81c27e09ec9bc38"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:27:40.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-23 11:18:12.000000000","tz":0},"subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","message":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027freezer-tempest-ipv6-only\u0027 job in freezer gate.\n\nDepends-On: https://review.opendev.org/#/c/682742\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: If9e8913e49d9ecf561a7e1824245a995fc6ac6db\nSigned-off-by: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a479b0421e12017d2f70b91e11061a7d5dcdb58d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a479b0421e12017d2f70b91e11061a7d5dcdb58d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"43d054e8a72e913eb7c13188f2bcb14e98d4cd1b":{"kind":"REWORK","_number":5,"created":"2025-11-23 11:18:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/682744/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/44/682744/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/44/682744/5"}}},"commit":{"parents":[{"commit":"e5a4d0cf75281bee076172dba81c27e09ec9bc38","subject":"Upgrade pylint to a new version that works with python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e5a4d0cf75281bee076172dba81c27e09ec9bc38"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:27:40.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-23 11:18:33.000000000","tz":0},"subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","message":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027freezer-tempest-ipv6-only\u0027 job in freezer gate.\n\nDepends-On: https://review.opendev.org/#/c/682742\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: If9e8913e49d9ecf561a7e1824245a995fc6ac6db\nSigned-off-by: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/43d054e8a72e913eb7c13188f2bcb14e98d4cd1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/43d054e8a72e913eb7c13188f2bcb14e98d4cd1b"}]},"branch":"refs/heads/master"},"2ca428db7d7a1bbdb5316ece0e7ff626680dc662":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-11-23 11:18:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/682744/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/44/682744/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/44/682744/6"}}},"commit":{"parents":[{"commit":"5943d3257f498fdfc4c0a830f03c2e76cc45adfe","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5943d3257f498fdfc4c0a830f03c2e76cc45adfe"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:27:40.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-23 11:18:47.000000000","tz":0},"subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","message":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027freezer-tempest-ipv6-only\u0027 job in freezer gate.\n\nDepends-On: https://review.opendev.org/#/c/682742\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: If9e8913e49d9ecf561a7e1824245a995fc6ac6db\nSigned-off-by: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/2ca428db7d7a1bbdb5316ece0e7ff626680dc662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/2ca428db7d7a1bbdb5316ece0e7ff626680dc662"}]},"branch":"refs/heads/master","description":"Rebase"},"b294b30892533c16c1a3235c801b18642c86f60e":{"kind":"REWORK","_number":7,"created":"2025-11-23 11:19:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/682744/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/44/682744/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/44/682744/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/44/682744/7"}}},"commit":{"parents":[{"commit":"5943d3257f498fdfc4c0a830f03c2e76cc45adfe","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5943d3257f498fdfc4c0a830f03c2e76cc45adfe"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-17 20:27:40.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-11-23 11:19:34.000000000","tz":0},"subject":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate","message":"[train][goal] Run \u0027freezer-tempest-ipv6-only\u0027 job in gate\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 which will\ndeploy services on IPv6.\n\nThis commit Run \u0027freezer-tempest-ipv6-only\u0027 job in freezer gate.\n\nDepends-On: https://review.opendev.org/#/c/682742\n\nStory: #2005477\nTask: #35890\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: If9e8913e49d9ecf561a7e1824245a995fc6ac6db\nSigned-off-by: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b294b30892533c16c1a3235c801b18642c86f60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b294b30892533c16c1a3235c801b18642c86f60e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
