)]}'
{"id":"openstack%2Ffreezer~691566","triplet_id":"openstack%2Ffreezer~master~Ibd0d363d24f28a104252fe09ec23473b0b9206d5","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Ibd0d363d24f28a104252fe09ec23473b0b9206d5","subject":"Stop testing python2.7","status":"ABANDONED","created":"2019-10-28 03:19:43.000000000","updated":"2024-12-03 16:34:35.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9bc2c3442de5379b1c685ddcb22bf0d3e0bae153","_number":691566,"virtual_id_number":691566,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-28 06:36:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2019-10-28 10:03:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 06:36:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 10:03:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"28846c5c1f30e6dff493a2b49e5967c0c30ca59b","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-28 03:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d8892fbe6fb94be2da9f08dc2c9d440204e4578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 04:10:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e12b662a4afe4252b7115ecd27fcbd41 : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a24770dfa3054f998251f01c30273b63 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e7fb6286fa84896aca64218b7fb75ee : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/442758f1dbb344cea373d3930cb36cdc : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f0297e4d404e879742865281cd3a9c : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b309593af154926b411733f1721afa2 : SUCCESS in 3m 19s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/83f754e34c9d4b08b361e6fd8b396102 : SUCCESS in 29m 41s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/d8e337ea56fe4e28be48d0946949bfab : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb0d77298e3a4e5ea2d4b3e20f0563a5 : SUCCESS in 3m 42s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/5922a2e5fb2744dcb7ec14990d8db254 : SUCCESS in 29m 38s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/7fce64c7399240f7899d2a3035694caa : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":1},{"id":"bb65ed6f2225a7c11ba1903f8e32890abb8a0d6a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 05:19:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"445740c348ea4898c575c526d491570502b8d9c3","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-28 05:55:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e42c17e226d6115e39d6713492a5991edabb6796","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-28 05:55:56.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nI switch it, please check again, thank you very much","accounts_in_message":[],"_revision_number":1},{"id":"b415be2142836b814f590e0e27aebb84980e9448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 06:36:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ed994e879644a3984e6c6065587dac7 : SUCCESS in 2m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4706c9870e924284af5381e7887919be : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff68e0f94bd4e0fa279f115638bc3e7 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a321330a60d4cc982711b39a43b70ee : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4ef962722e7485495e56d35912d4eb4 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99089ffe41ed4065ae45f13ab0dd59ee : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/617041ce5fdc4ca092042a93494783e6 : SUCCESS in 2m 59s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/55cd5f1152144d0893181732b174001d : SUCCESS in 31m 10s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/3aad6a7c393046c9b862dc6bef134a1b : SUCCESS in 39m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e6795700c2784b0ba4b4b8cc54989e0e : SUCCESS in 4m 07s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/2a1cdd1ef8f74200a584e99eefdaf836 : SUCCESS in 37m 35s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/6cd12b8c9f7d4abd87c0ac4105be6f77 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fcfb71e7c10d30c08e03d68afc22b2389a97a281","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 10:01:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ffc2e09625e64491b2c385f101cff4edde87972","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-28 10:03:56.000000000","message":"Patch Set 2: -Code-Review\n\nPlease follow https://review.opendev.org/#/c/691178/4/goals/selected/ussuri/drop-py27.rst","accounts_in_message":[],"_revision_number":2},{"id":"9bc2c3442de5379b1c685ddcb22bf0d3e0bae153","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-03 16:34:35.000000000","message":"Abandoned\n\nhas been implemented with different patch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30a0077221599d4874d6ed05c279a34699e99069","revisions":{"d1c4f99f051dc77f4bbb333c7cedd7246f28bbfd":{"kind":"REWORK","_number":1,"created":"2019-10-28 03:19:43.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/66/691566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/66/691566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/691566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/691566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/691566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/66/691566/1"}}},"commit":{"parents":[{"commit":"9d9294fba4698d458857ec1fd814c13079eb2da9","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9d9294fba4698d458857ec1fd814c13079eb2da9"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 03:19:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 03:19:33.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: Ibd0d363d24f28a104252fe09ec23473b0b9206d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d1c4f99f051dc77f4bbb333c7cedd7246f28bbfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d1c4f99f051dc77f4bbb333c7cedd7246f28bbfd"}]},"branch":"refs/heads/master"},"30a0077221599d4874d6ed05c279a34699e99069":{"kind":"REWORK","_number":2,"created":"2019-10-28 05:55:19.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/66/691566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/66/691566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/691566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/691566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/66/691566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/66/691566/2"}}},"commit":{"parents":[{"commit":"9d9294fba4698d458857ec1fd814c13079eb2da9","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9d9294fba4698d458857ec1fd814c13079eb2da9"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 03:19:33.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 05:55:02.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: Ibd0d363d24f28a104252fe09ec23473b0b9206d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/30a0077221599d4874d6ed05c279a34699e99069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/30a0077221599d4874d6ed05c279a34699e99069"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
