)]}'
{"id":"openstack%2Ffreezer~692254","triplet_id":"openstack%2Ffreezer~master~Id5f91aa5dd22b1364fe35143b6c96e834284d252","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Id5f91aa5dd22b1364fe35143b6c96e834284d252","subject":"Upgrade pylint to a new version that works with python3.7","status":"MERGED","created":"2019-10-31 02:41:15.000000000","updated":"2019-11-01 05:10:05.000000000","submitted":"2019-11-01 05:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692254-1572584921380-8bdcbf74","meta_rev_id":"7727ea0e6022ff5002b1c15bb368f78f4f4f4f74","_number":692254,"virtual_id_number":692254,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-01 05:08:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-01 04:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-01 04:28:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-01 05:08:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05dd56e760b80108f621ce13818d0efa7e536cef","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-10-31 02:41:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"252ba280d0e226eeec4795663eb48b33c496dfe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 03:36:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5bf715626aa4419387ade0bfd9c4a9da : SUCCESS in 2m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4260d149abdc4aaf8cec94c68bff42c0 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f8bccd211d44b9fb16bdf352b894bb7 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/453c6c176c954d1ead5cdfd23cb13c5a : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b94ed8ce44d4442b7479bb9da51ff25 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53156814a10f416b8e07185b5b9b8bf3 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/908d9a44de314a1c879dc6516a3c70e3 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd96acb838c3469bb4ba498daeff174e : SUCCESS in 3m 40s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/fc1fa51bf6384abe8d87a5766d4dd32d : SUCCESS in 38m 56s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/9ba122e4348c41dcbe0b01169595e8af : SUCCESS in 45m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f49434d3914641e998b00ba74538bf08 : SUCCESS in 4m 10s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/41a266d17ca44623b9940e0409e8a898 : SUCCESS in 40m 16s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/f9f44479e957440eb9d18a9fdb4eafa0 : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0fba908b2c6e5155ebb236941b37e0a885d04bcd","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-01 04:28:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e631d26a957720b6be38cfe182ad7e64da44c377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 04:28:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"576f9c190d3c9588fc89950cd19a55c108650e6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 05:08:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/670c87bc275f4ed6bb20b8d129297755 : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63355cdf58474acfbc38eb5f2da8dd59 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34e6d38b20047a996cb01e22b2cbf24 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b156bfcf963a4ebf94597c2b78ce19af : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/360704e427674d719c74325bb1de7b02 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7a981fede7841589dc94fcc9a43e3e9 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ed06faf78e42af8d5b2083b66420fd : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cfbd130d2ab40518a219e715b0e4ad7 : SUCCESS in 2m 54s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/05ee936f1830498eb66b516190923dca : SUCCESS in 34m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfbd7484ae2e4b78bae91886333ae700 : SUCCESS in 5m 15s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/9f002d7a9bba4dec85c252aa835da880 : SUCCESS in 34m 29s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ed49c7ed822c497c9e7439cef680ca35 : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":1},{"id":"25c091df0351982016103c53ecae0925433bb42c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 05:08:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7727ea0e6022ff5002b1c15bb368f78f4f4f4f74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 05:10:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e314b9586d5b4fcdb96ea3cb3be6c26a : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3e11312535f4e5f854e81b608076f24 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5a4d0cf75281bee076172dba81c27e09ec9bc38","revisions":{"e5a4d0cf75281bee076172dba81c27e09ec9bc38":{"kind":"REWORK","_number":1,"created":"2019-10-31 02:41:15.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/54/692254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/54/692254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/692254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/692254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/54/692254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/54/692254/1"}}},"commit":{"parents":[{"commit":"497d1af6533c706cf50b34e74c1db71c412a7873","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/497d1af6533c706cf50b34e74c1db71c412a7873"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-10-31 02:33:54.000000000","tz":-240},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-10-31 02:41:02.000000000","tz":-240},"subject":"Upgrade pylint to a new version that works with python3.7","message":"Upgrade pylint to a new version that works with python3.7\n\n1.The pylint \u0026 astroid verison should be specified for python3.7\npylint is upgrade 2.3.1\nastroid is upgrade 2.2.0\n\n2.Some rules are added to the ignore list in .pylintrc\n\nChange-Id: Id5f91aa5dd22b1364fe35143b6c96e834284d252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e5a4d0cf75281bee076172dba81c27e09ec9bc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e5a4d0cf75281bee076172dba81c27e09ec9bc38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
