)]}'
{"id":"openstack%2Ffreezer~692778","triplet_id":"openstack%2Ffreezer~master~I1935268db988aa22a6d9f146214e93b56f35143a","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"I1935268db988aa22a6d9f146214e93b56f35143a","subject":"Update tox.ini","status":"ABANDONED","created":"2019-11-04 09:12:01.000000000","updated":"2019-11-18 12:20:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5fffc77d80dcbcf53c052ce368820995fa885622","_number":692778,"virtual_id_number":692778,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-13 06:30:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-13 06:37:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 06:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f145e2c9260565fe1504cbd7a8f7f53385e2722","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-04 09:12:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55307e2d75460dd42ea9a8b08484954f7383fb76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 10:02:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cde211e322e4662ad7587147763ed98 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d9895469abf4815bfa45200609da001 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10b257df10904a998c2584f5951741fa : FAILURE in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8650b0fdf02745c3b70d5cd262e26d73 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/868908e838774a538a7e404b5d69d018 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dcd19b2788f4c61882eb28ec3e33feb : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a70b4109b3c648a6bfd58a28ecae447c : SUCCESS in 4m 06s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/3e4ad25d83404f77b898e045af0be1ae : SUCCESS in 33m 06s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/b05e564561a64eab8fbd1e0849857205 : SUCCESS in 48m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c77fe90c59f644adbcb0c6f01c27d18d : SUCCESS in 4m 25s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/45a0af568d4442aeb822d34e0d86fda1 : SUCCESS in 32m 12s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/9084cd4ae0714f53b04c67a9dd7b3483 : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c9865161b60263bbaea4bad92950a056b78e66f1","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-13 03:31:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78c5b64a26d6960cab6424ba0192b18d638605a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 04:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad1d8f0d0de844798375f94f615030ea : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c76d9707f54944ba202b16fdd8861a : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67f76a27954e48cb904142bda12562f7 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fc2945c750a454fb144ebc14dbf06e1 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d37c22ff94184f19a6bc4d62e14cf439 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c97e070ef84269ac8a3cc0280632d2 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/914e7ab2f57e4697a71bf74fa315e10c : SUCCESS in 3m 01s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/e192b536637647b78d18d2207cf1975d : SUCCESS in 36m 36s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/cc11bd9bb6e04050ac49e1df8e901330 : SUCCESS in 45m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0a3d54424c604d719d149d4f392b36de : SUCCESS in 6m 04s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a4815ac79e16413497329bef189958ae : SUCCESS in 32m 10s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/136c5238d6df44c7b3b220e051433f01 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2873ca382d3748e9a4ea91a6dd6fa3f2a7b3b75b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 04:19:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"385cfa5cb6d640e54698581d3686ea842b96b399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 04:58:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc0cd8797bb04b4b9b6cdb76220ebf18 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c7d80b1a7841ec9d49c93c43ffa759 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64b2167e1c2c4ff396d4c1b7100c4100 : FAILURE in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf63a62ec7140c7adb3b11afd82034e : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d39aa1f525db499682b17ca2c78ca341 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29a333fc83040129110e76a4b31761c : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fda796b19c64e4b8ee831d863dd6c0d : SUCCESS in 3m 18s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/0120ee7c38644b55a46126442ae12c71 : SUCCESS in 37m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1af1e7eee3fb4d64897cfbe2c7465e4f : SUCCESS in 4m 12s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b305434c52394023bf9e8c79eb4c19d3 : SUCCESS in 33m 25s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/2e37024fa39b49639c529e643e85372c : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fe34486921f86d68c94d85e8f65ff96d9a0f210a","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-13 05:50:30.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"9da0b91b165abcb0d11e17b595051741e641ea7a","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-13 05:51:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6014e2a8fa36c518c9f1af9e5129f62df456037e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 06:30:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1b6e9fa03c7414eab9bc9f2f0ffc6d8 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba4849c06f746f4ad7dae75dadeaa3b : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b65b3af2eed4520bced8873bf014768 : FAILURE in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/194f3770a5804dd4953fd3056d453b29 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f170bd74fdb44f08a49ac304cfdce58 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a63e8fe420c48df8772b14156a52698 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84bbeaf4f96f498cb8f22e6cf2b3c5c9 : SUCCESS in 3m 10s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/76184691c54d4147b240a938bc530a23 : SUCCESS in 31m 48s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/5f0f8f85a4ad4333a503d347832fc506 : SUCCESS in 38m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4628ab7e6e44101a9a2913dea9e70db : SUCCESS in 4m 24s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/cf2a3f2fec374a7580cb11da74e02d9c : SUCCESS in 38m 03s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/0d5cc10d721b4c0392f80e1b8fa05a34 : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":1},{"id":"98f1c8ecbd61fd6ac57caff67de67b209599335a","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-13 06:37:25.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"5fffc77d80dcbcf53c052ce368820995fa885622","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2019-11-18 12:20:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f7888ad0745e7f57105ea970a9b5b6f900608d1","revisions":{"1f7888ad0745e7f57105ea970a9b5b6f900608d1":{"kind":"REWORK","_number":1,"created":"2019-11-04 09:12:01.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/78/692778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/78/692778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/78/692778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/78/692778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/78/692778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/78/692778/1"}}},"commit":{"parents":[{"commit":"e5a4d0cf75281bee076172dba81c27e09ec9bc38","subject":"Upgrade pylint to a new version that works with python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e5a4d0cf75281bee076172dba81c27e09ec9bc38"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-04 08:53:50.000000000","tz":480},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2019-11-04 09:11:42.000000000","tz":480},"subject":"Update tox.ini","message":"Update tox.ini\n\nAdd py36 in envlist.\n\nChange-Id: I1935268db988aa22a6d9f146214e93b56f35143a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/1f7888ad0745e7f57105ea970a9b5b6f900608d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/1f7888ad0745e7f57105ea970a9b5b6f900608d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
