)]}'
{"id":"openstack%2Ffreezer~695615","triplet_id":"openstack%2Ffreezer~master~I14ab74f3e82e83ef5c5a49831f68335965e7e101","project":"openstack/freezer","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I14ab74f3e82e83ef5c5a49831f68335965e7e101","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-22 01:53:15.000000000","updated":"2020-01-16 08:36:39.000000000","submitted":"2020-01-16 08:34:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695615-1579163654311-af565c7f","meta_rev_id":"1d287b5521d05c23b0a340d083f77dcedb723bfa","_number":695615,"virtual_id_number":695615,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-16 02:41:40.000000000","_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"date":"2019-12-13 00:24:15.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 08:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-16 07:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 07:50:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 02:41:40.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-01-16 07:50:47.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-01-16 08:34:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f2064a456d4422c4003c8fe0c50af093bd08eb9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-22 01:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"153eac2fc07426d53bdf8cbf46b0ea6c84322bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 02:26:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9f50129d66f2406d8072e8ff90481a61 : SUCCESS in 2m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0822053709bf4e27a7348e1addb81568 : SUCCESS in 2m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aea85c476ab439883d333c2593b920b : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db77a91ca744d4facb07f8f9dd4ef0c : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88bec33f028549088a0fe223ecc96cbf : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/530b18cb9644479c979f5633b2dec997 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b0842bd16c400189ebc4b636b3f367 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c3c71e24e254ec89779e8ed2c6da678 : SUCCESS in 3m 31s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/5a28ef5cfe6447c9b468979e47c295f6 : FAILURE in 29m 33s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/5b479d222ebf451bb04774577151a957 : FAILURE in 9m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/338fbea805f64e3d8c8dff73013a6f63 : SUCCESS in 6m 07s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/47225878d67b4c5e89d855e298531e77 : FAILURE in 32m 52s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b75f7f51aa4f4a989cf88526db30a38c : SUCCESS in 28m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3e49f4a989948c70a40cb68ad655d189bde1c0ac","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-13 00:24:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0bf57c70069c3a0fc9a2325050d6fe32f760283c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 01:05:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36733e7c7aec474a9a941fee1aa7a7f3 : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0b1336d07af4dfd86ff26d190d77161 : SUCCESS in 2m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ad36fcc02d14cd18138e88be5b9019c : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f80967f5004aa59d6054ed1fb707fe : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4dec36e786346cfa35c35fe0e67835c : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16fad363dda74d6e91521b427eca0c6d : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986865549d1c4fac997921a7ea75521e : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11d5e7b2531943e18416bdb8cbed9b5a : SUCCESS in 3m 33s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b22cd5241baa4d2c8fa11958cc70d8e2 : FAILURE in 35m 34s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/b84e6ce1218246a688541ab0476f2443 : FAILURE in 11m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eeabbee5c82f4b5aa0a8dbaf266669b6 : SUCCESS in 4m 19s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/dfd1cfb916f742d58fd9ecb6c18f65ef : FAILURE in 37m 21s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/52738c31d7aa430faa5646da8ae024bf : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c163018130382414c0818cac43d2e6f77ac6a3b1","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-01-16 02:41:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ffda5bf91b10bf62065c8efa27e10fe12fcdb60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 03:24:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac102f70427f4f8dbeb4860b04c0bc2f : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8b9c399a4f54b418a72d926a3f47771 : SUCCESS in 2m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2403f71ac724790bf0038f0d6f42900 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ded81615d7c437f92db3e4cc59567cd : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a55629a58f1b42d6a05b227ef3de1416 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5290e5ef7e12434e9e499ea34d38c2a8 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986915043e66433da829f33e8c8f87a7 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/703ad9af465142cf9daa7b61ffce101a : SUCCESS in 3m 28s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/55fb7ffb7ea34e45804f06550ecc8366 : FAILURE in 32m 17s (non-voting)\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/e30eae25b6c54d4f917acf708cc0706c : FAILURE in 10m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f55af7c452fc4865a5c4d151eba47846 : SUCCESS in 4m 02s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b0e05d2bc5ee4008a6a7af12ef1799b5 : FAILURE in 29m 39s (non-voting)\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/759fa358e0924213a12473bf6a876ae7 : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":1},{"id":"17bd54616f5dcb4f5cf88ef8f1f7500d93430831","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-01-16 07:50:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM.\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"3299b19a181debdc09f58643018baf9c7573ab69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:51:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36ffcc0bbc4dd2b6889157e58cdf343868520c8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:34:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d93b9f187f28427594ae8c71ef72b741 : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/66768ff33d2b42e0b0ab98f793988b3e : SUCCESS in 2m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f6a2e425cb949a2a3c7409dc31356f4 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551d217166814c1693e15be320f5cf0a : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2983bc44f1c418ba51ddd002485825f : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dd7ded136794038878f656416b805b8 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809241b8abe746e8841afd70c48d27f7 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6ad71a2c8a1464fb1cb4044f58b114d : SUCCESS in 4m 03s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c36fe7ac0cca45f393452d5f2b902c17 : SUCCESS in 38m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10a4e5c71117415cb9bedb7ef076e745 : SUCCESS in 4m 47s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b5dc71b27558428ba5e80540a8726b81 : SUCCESS in 34m 40s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a59b875e0b6240ce9ad2f002d1322527 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":1},{"id":"648663f26138706648737357d62c6d76a6345ec6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:34:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a9a44186145ad50169a98728b4f17ed6af4d1ddc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:36:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0685eda8bdb746199c3063b0df19132a : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ed44e387374c889954279869510340 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f04df2caa58717dc1667d02c86f2da8d6a261c08","revisions":{"f04df2caa58717dc1667d02c86f2da8d6a261c08":{"kind":"REWORK","_number":1,"created":"2019-11-22 01:53:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/15/695615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/695615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/695615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/695615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/695615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/695615/1"}}},"commit":{"parents":[{"commit":"3ffcc0c657f0303081f0e087b932f6348335340a","subject":"Update devstack doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3ffcc0c657f0303081f0e087b932f6348335340a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-22 01:52:55.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-22 01:52:55.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nfreezer is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I14ab74f3e82e83ef5c5a49831f68335965e7e101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f04df2caa58717dc1667d02c86f2da8d6a261c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f04df2caa58717dc1667d02c86f2da8d6a261c08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
