)]}'
{"id":"openstack%2Ffreezer~698534","triplet_id":"openstack%2Ffreezer~master~Id60e75b5f999b70626addaf47b4abd1ad15555d6","project":"openstack/freezer","branch":"master","topic":"fix-configparser-boolean-values","hashtags":[],"change_id":"Id60e75b5f999b70626addaf47b4abd1ad15555d6","subject":"Fix parsing config file for freezer job","status":"MERGED","created":"2019-12-11 16:42:14.000000000","updated":"2020-01-17 01:41:45.000000000","submitted":"2020-01-17 01:40:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698534-1579225229551-5f1f21a0","meta_rev_id":"063a3d38ddf188f6a43377a89e29c37712ec593d","_number":698534,"virtual_id_number":698534,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-17 01:40:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2020-01-17 00:56:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2020-01-17 00:56:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 00:56:59.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-01-17 01:40:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54e872aecaaf2a116c38757c2f769fc965abed57","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-12-11 16:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6035e9e2a652108ef67c6d65cec71cd751443b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:31:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/407a3b8802524cfbba8b000b2430ccc2 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b27937b04049b59e9178fd75fb322b : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05a9c784eb3b4521a15a3488d1af988b : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b72783061ae4c76baaf0dded2c66171 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1ee0d1671de4e9688c45068d2a7bbaf : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d842f0f89e2a4e4995b30cc10012ca2f : SUCCESS in 4m 40s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/a6bc47b390f143e79298c43b40731a2d : FAILURE in 46m 10s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/f2cbc121cb9f49139c2c88510a270132 : FAILURE in 20m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/974edc3735ad47e9b57f5945385b0740 : SUCCESS in 7m 28s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/3848e962a4d649558f5532b10260319d : FAILURE in 38m 27s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/36210f34361f45c0942a0fdcbbbd68a5 : SUCCESS in 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ccff8ba5755f8b253552bff4996ff4a89a101208","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2019-12-21 02:49:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f68c984bdd5f7a096ba9b77cbb655e866aef1fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 03:33:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f92453ea17745fa9f84995e44882cee : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd64a610108408a9439412a60cbd189 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16cbc6c005e54a8baf0816ed1b51a595 : FAILURE in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95decc0db2a947bf94321e461c56aea2 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44623bb124764b00bfc76756f6a3ea28 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef02fea5265e4fcd823afad5390889b6 : SUCCESS in 3m 56s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/d435e736a60a4141ad9f07479a4703f8 : FAILURE in 36m 33s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/ba7c94f6837446f696a49d4992439fc8 : FAILURE in 14m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22f97d17a37243f99b9ccf0729b062c5 : SUCCESS in 4m 10s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/8e093b9ff8a34161a626a75f337d1eb2 : FAILURE in 34m 20s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/9cc4a1f9bd164bd2941352b23bfdf514 : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"66dbc022f64671669e848ea32d6da474587d940f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-01-02 10:26:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09be6a1ef02698fa23c73f3c438199d291a7f8f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 11:14:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75a8e8e2a43d4e62a0c22bf2cb796ebd : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f69c8e6528104cccb36781f74c563907 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69befb716fed42f590d1d4dce4444233 : RETRY_LIMIT in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d71abe308244133ab4d2d0bfdb22f7c : RETRY_LIMIT in 2m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9ec5c83d71e4adbb30ff9321c66d131 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787202e4f08947b887ffbda5c6a71c65 : FAILURE in 4m 30s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/f6f0386a56604c74be51f035782bfc6c : FAILURE in 36m 50s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/1f6c77251c464facbd5f86ca659177d3 : FAILURE in 15m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37e38242b38e429294f5f203ef056b4d : SUCCESS in 4m 24s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a2f211f802d045688597310c1ce7a216 : FAILURE in 41m 15s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/00d6b88da03040078eb4bff029763074 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0405c707294621fb121b9437ac68b1a95ed7d504","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-01-16 11:40:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e2541a6a90b2af4da600025b8860b0829e16f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 12:23:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ae1888019924cbf9c95bbad442c1c3a : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a508ec91327d4897a4f094c24bba7782 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cce64a27d33547279cd6501263b4fc6b : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38f6406327ee4210b01e55c62e83cc75 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dcf13aca7354bd0bb43377ad0763ede : SUCCESS in 6m 18s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/91affa983eab41c9a24f4e4548b9398f : SUCCESS in 38m 30s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/05e83faa360446fe9102a3f642505443 : FAILURE in 12m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0239d103f4b241bd94d8e221e106c8eb : SUCCESS in 5m 30s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/72cf066bc817477d8c183a02d0cea05f : SUCCESS in 37m 33s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/9c642f280f1046ff90816e2bbc942d82 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":2},{"id":"5f34e554d2924ad6718fdbcca20b4e52c0c8c1c2","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-01-17 00:56:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"f2c03cdd86a28d502391c5aa0fe20eda1f4f3291","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 00:57:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c82c2b04cc7a13cd9ed0078880dc7ea37a58171","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 01:40:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"12ac33035a9f55bf373d7b3e08ca9b1b126ee3f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 01:40:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/accc7dee633049aa874b6678bfc6b21e : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ba93dd92854994abdf024b11c82d8b : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7fb04c82ae0440a90562b1ddf0d863c : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27b3c5df385f4562aa68737230b3f0ee : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb67d39e783046bf8ffb5b2792d9d63a : SUCCESS in 4m 44s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b499abdfbbb74b44b18290677d29a7d3 : SUCCESS in 41m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e526ba6e84514132a3620204ee9a3a3c : SUCCESS in 4m 54s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/422d6ada2d9346cbb721bd28f5d7791c : SUCCESS in 37m 43s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1a131c42b7434b1e861ceeadb719bff6 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":2},{"id":"813360b456b5161cb4facb268eff6ebb92e03edf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 01:41:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d9f141a52a4a64903e1930d0b34124 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bcaac1e3c1e1951ac37ce65b68b5920a6bca6911","revisions":{"ecb7df291f8b01f2b1310f3575e5400c7877b7f1":{"kind":"REWORK","_number":1,"created":"2019-12-11 16:42:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/698534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/34/698534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/698534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/698534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/698534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/34/698534/1"}}},"commit":{"parents":[{"commit":"3ffcc0c657f0303081f0e087b932f6348335340a","subject":"Update devstack doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3ffcc0c657f0303081f0e087b932f6348335340a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-12-11 16:38:47.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-12-11 16:38:47.000000000","tz":60},"subject":"Fix parsing config file for freezer job","message":"Fix parsing config file for freezer job\n\nConfig parsers do not guess datatypes of values\nin configuration files, always storing them internally\nas strings. This means that we need to convert booleans, as\nthey are parsed as Strings from config file.\n\nChange-Id: Id60e75b5f999b70626addaf47b4abd1ad15555d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ecb7df291f8b01f2b1310f3575e5400c7877b7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ecb7df291f8b01f2b1310f3575e5400c7877b7f1"}]},"branch":"refs/heads/master"},"bcaac1e3c1e1951ac37ce65b68b5920a6bca6911":{"kind":"REWORK","_number":2,"created":"2020-01-16 11:40:16.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/698534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/34/698534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/698534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/698534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/698534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/34/698534/2"}}},"commit":{"parents":[{"commit":"4ff76740294315508c6bd5b746c1b4ac2fc33dc9","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/4ff76740294315508c6bd5b746c1b4ac2fc33dc9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-12-11 16:38:47.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-01-16 11:37:59.000000000","tz":60},"subject":"Fix parsing config file for freezer job","message":"Fix parsing config file for freezer job\n\nConfig parsers do not guess datatypes of values\nin configuration files, always storing them internally\nas strings. This means that we need to convert booleans, as\nthey are parsed as Strings from config file.\n\nThis patch is fixing typo from my previous commit where\nNone is needed to remove from second if clause as it is\nalready in first if clause and option_value should be False.\n\nChange-Id: Id60e75b5f999b70626addaf47b4abd1ad15555d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bcaac1e3c1e1951ac37ce65b68b5920a6bca6911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bcaac1e3c1e1951ac37ce65b68b5920a6bca6911"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
