)]}'
{"id":"openstack%2Ffreezer~701006","triplet_id":"openstack%2Ffreezer~master~I2a356379c830156c8aba32d137ebd12a3ad49b31","project":"openstack/freezer","branch":"master","topic":"log_format","hashtags":[],"change_id":"I2a356379c830156c8aba32d137ebd12a3ad49b31","subject":"Fix missing parameter backup_id in the log message","status":"MERGED","created":"2020-01-03 09:48:24.000000000","updated":"2020-01-15 08:50:57.000000000","submitted":"2020-01-15 08:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701006-1579078150616-1f963bb9","meta_rev_id":"2fe37b1b7bb285c78236f24b6e7552995a06666f","_number":701006,"virtual_id_number":701006,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 08:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"date":"2020-01-15 02:13:39.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2020-01-15 08:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-09 07:42:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2020-01-15 08:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 07:42:34.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2020-01-15 08:10:57.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-01-15 08:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3755248095037cf4ea2c8498dd9dcd4ff70f498a","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-03 09:48:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49279474e331a282996491aaac829fb439620e54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 10:32:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63cdb1c66c0f40f988f8f13c5aa3afd3 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88ce672a9144e65b4bceda3ef389a5e : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdb198b8246f4421bb7827c71b5416a6 : FAILURE in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78aaf6d7a3e54792abcb6ee9f54f0e5a : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a24f494c9a1b406f99b86a633cf49935 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be386f9ae1b54662a323545235b4215e : SUCCESS in 4m 40s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/2fd65a9148254507b07ac63165df0729 : FAILURE in 36m 54s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/d3f6b32a9149434ba3cb649b57ada984 : FAILURE in 15m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/72b7ae74a8c24efc99ad99d6639e46df : SUCCESS in 5m 19s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f9551ca594f64c3ea0539cf69ad91eb2 : FAILURE in 35m 11s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/54a6ecf17ec645bfaf205975d14f7ae9 : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":1},{"id":"75db2c18e9ab3c40ae476375bd1c8c8bc5f6d4d1","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2020-01-09 07:42:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0084fb25ddf818acf8f882af74dabdc19aaf756b","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-13 00:58:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0255402d8f810d8ad4d88171a94a6dd16f37a5fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 01:39:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62f8f9ed63414d819a755aa5a2ded725 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c8a1def71c43d8ae51eabc54a2470d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d883daf5c74e480186845aa21bd36cb3 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9755b4bc19d4a088743097714ef826b : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3902b5d13541ca956953882885e2a0 : SUCCESS in 4m 41s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/0e673f81d7194adb8f0cc0ae647ee01f : FAILURE in 9m 27s (non-voting)\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/bf7c80945c4f4828ab3bb908432d7688 : FAILURE in 12m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35d15f40f3ad4b108fb80061ff625286 : SUCCESS in 5m 51s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/5537ff93ccba426a930bb6bfe148283d : FAILURE in 8m 59s (non-voting)\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a726ed6d85f840e69947d1858cfafad6 : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c3d00e4118ae8cb59632991fb2593be02e1ce8b1","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-15 02:13:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22cc9fa59b5780370fa1cfba76132ce8b3c534c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 03:21:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71f15c13e5994a1d814991b3aa3b0f7b : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04acf0ffab824f669eace7f1bcc8312d : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af81517f264f43768acf291a8be6cd68 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/779c340b19964502b03cdae850974043 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf1f1a5871c4b598561dbc953ba9d4e : SUCCESS in 5m 54s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/7994e19b3442407fa81e146228138a6a : FAILURE in 52m 54s (non-voting)\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/c748ccb2ade7420e98c5570344d0bdd4 : FAILURE in 17m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63369d6d45a04eaeb4818bfced76df29 : SUCCESS in 5m 21s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/0d191542c1964b8e879d0945dabcc47d : FAILURE in 57m 33s (non-voting)\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b4e6f21b529c4d658ebf1187a2375860 : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0f61a70f509c03638b4fa2aa7f56bc3798322be2","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-01-15 08:10:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3e708d8d18e30411cc9e709b239cfe6f0f9260a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 08:11:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ef066bb2b4c00103889dd0094b60cd9c057089c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 08:49:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52df0fba5343d1a6e506a2ffb61cbdc59c38a666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 08:49:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e43258e428f4afd91683a524969ccbd : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f07103f153934a39aeb9113572f4d875 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df96be6bf4aa41be8b66548d66697346 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e5c53833edc41849fda3c60ac161314 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb38a0d5cd1420a90bc8610620d5af3 : SUCCESS in 4m 38s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/254b26718d2f4d5188eeb473051b6f84 : FAILURE in 35m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/65c6a52f654f46ca8038994a3c80ec77 : SUCCESS in 5m 42s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/1d95bb98d7114d738d6583ca194b36de : FAILURE in 33m 08s (non-voting)\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a46146a7dc184892bd4611e03fdb9e85 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4ca34fb8c9f97c730d00166ab902bfac598e82bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 08:50:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2625a8bff0e249d59596973c17639452 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a9bcfb9076cd9996cd48c53e228cfc772936d9a","revisions":{"7a9bcfb9076cd9996cd48c53e228cfc772936d9a":{"kind":"REWORK","_number":1,"created":"2020-01-03 09:48:24.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/06/701006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/06/701006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/06/701006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/06/701006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/06/701006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/06/701006/1"}}},"commit":{"parents":[{"commit":"3ffcc0c657f0303081f0e087b932f6348335340a","subject":"Update devstack doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3ffcc0c657f0303081f0e087b932f6348335340a"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-03 09:24:21.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-03 09:24:21.000000000","tz":480},"subject":"Fix missing parameter backup_id in the log message","message":"Fix missing parameter backup_id in the log message\n\nThis is to add missing parameter backup_id in the log message.\n\nChange-Id: I2a356379c830156c8aba32d137ebd12a3ad49b31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7a9bcfb9076cd9996cd48c53e228cfc772936d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7a9bcfb9076cd9996cd48c53e228cfc772936d9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
