)]}'
{"id":"openstack%2Ffreezer~702799","triplet_id":"openstack%2Ffreezer~master~If341ec109bff4daef37347ef65cd3c903f74fba7","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"If341ec109bff4daef37347ef65cd3c903f74fba7","subject":"Update some tempest jobs to voting job","status":"MERGED","created":"2020-01-16 05:42:01.000000000","updated":"2020-01-16 07:48:11.000000000","submitted":"2020-01-16 07:45:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702799-1579160718398-bbef5f2b","meta_rev_id":"8a0ab3acfc8209d8bbad7819c361f4fe05532c98","_number":702799,"virtual_id_number":702799,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 07:45:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-16 06:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 06:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 07:45:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94d9d4794b84e3b0166a5a9bdefb4b1f5222c282","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-01-16 05:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bb6ead8f537e903974573efc602356836d03500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 06:29:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e0ffb03a4e0492ca15cc8e0f6850bd6 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/605855ddc2074c57b22e4da422b74e07 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eb6908bfac04f7bb659bad207463740 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bfa931e1ccd48b2b5ef8079d4b99b26 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91997db610b489a942aa187a70846e5 : SUCCESS in 3m 40s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b8b3f70cf56e413a98239909fbf00cd0 : SUCCESS in 41m 47s\n- freezer-centos-7 https://zuul.opendev.org/t/openstack/build/8b86914ac8074e94a040fa4ad1fe3fe8 : FAILURE in 17m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db1fb7bc0e224dceb99fa87aa88f738b : SUCCESS in 4m 42s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a3bbe8d844bf48cfad7394548105e9c2 : SUCCESS in 40m 07s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/0ed78643f6f94dfbbf68ff41416c2fce : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c38d909cdf324a3cb22770eb73ab798f9ff10bab","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-01-16 06:34:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8710aaa651850474526f8ffbf1ea72be679830c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 06:34:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6fe47a1017b0ff3a07b0884c64e755918557e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:45:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93ed43d2de0342a89ada36fb0453fafd : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df9e042b0e54f88ae6c4d6fea6a6494 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88f846e9427541cd9db8cfc58b32e8e7 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cee9b061903e4991a45123bab66e94df : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2c282bb54e43ebbc997782a46e895c : SUCCESS in 7m 05s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/f5d347996b1f4cc297103ba6476ceddd : SUCCESS in 1h 02m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58da7ca70e404873a8f2d26385609e50 : SUCCESS in 8m 29s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/8c29f38e08cc466784a76b42fcefe22c : SUCCESS in 45m 09s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/be74b65afa984ec4936aa4c58103cc5c : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bb3340de03d3d8805eaba7b5d15d3412ed807d80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:45:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a0ab3acfc8209d8bbad7819c361f4fe05532c98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 07:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd3a8e06fb94ff8a0bbd7d2b537b25b : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d39ec438c2214ac2ea33b0f7e82126f97938a65","revisions":{"6d39ec438c2214ac2ea33b0f7e82126f97938a65":{"kind":"REWORK","_number":1,"created":"2020-01-16 05:42:01.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/99/702799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/99/702799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/99/702799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/99/702799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/99/702799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/99/702799/1"}}},"commit":{"parents":[{"commit":"c1231d0ebb7909c9b101ec68dcf2bce8c3b7793f","subject":"[Trivial]Fix typo of instnace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c1231d0ebb7909c9b101ec68dcf2bce8c3b7793f"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-01-16 04:59:40.000000000","tz":-480},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-01-16 04:59:40.000000000","tz":-480},"subject":"Update some tempest jobs to voting job","message":"Update some tempest jobs to voting job\n\nUpdate some tempest jobs to voting job.\n\nChange-Id: If341ec109bff4daef37347ef65cd3c903f74fba7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6d39ec438c2214ac2ea33b0f7e82126f97938a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6d39ec438c2214ac2ea33b0f7e82126f97938a65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
