)]}'
{"id":"openstack%2Ffreezer~721915","triplet_id":"openstack%2Ffreezer~master~Id34ad306a28f77785abf2c84f16a091658558bff","project":"openstack/freezer","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"Id34ad306a28f77785abf2c84f16a091658558bff","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-22 13:09:56.000000000","updated":"2020-05-09 02:27:56.000000000","submitted":"2020-05-09 02:25:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721915-1588991109054-4f1731f4","meta_rev_id":"4e128f80c0cd9da1852cc94c71ce52bca92d5975","_number":721915,"virtual_id_number":721915,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-09 02:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"recommended":{"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},"all":[{"value":1,"date":"2020-05-08 12:20:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-05-08 02:51:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},{"value":2,"date":"2020-05-07 01:50:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-26 09:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},{"value":1,"date":"2020-05-07 01:50:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-06 01:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 01:00:21.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-05-07 01:50:24.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-05-08 02:51:41.000000000","updated_by":{"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},"reviewer":{"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},"state":"REVIEWER"},{"updated":"2020-05-08 12:20:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-09 02:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62b31460ad8d2aad1484ac754fabd97183c07cd3","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-22 13:09:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9a9126dac3a0cc9eea5dc435cce2401f1e69dc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 14:27:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1bee342747745fc953a0c1bee56cfb1 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec9771922b9479db8e65bb3c94790e0 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dfbaf74ce244db9a8a95a58dc43d6f2 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5714a94cbb69415e98e15d0580c4ef22 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3e39e368a344ff3a712f014395c2aea : SUCCESS in 7m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2eae8113ff04f2ca26c75b13152c585 : SUCCESS in 6m 10s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/344a7d6cb1304da68c71458281be625d : SUCCESS in 37m 50s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/7023aea28dde4467b0f57936d395c419 : SUCCESS in 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b139366b8e554daf54d466781b24b48f0e658f07","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-04-26 09:13:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32da03b38c77052e55f46cf096269936f13f6807","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-04-28 01:55:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa94ae05314812ea8b09e4376c089dc3ad42402b","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-04-30 09:01:26.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"4e3a8206b5a2129fc968b04f8b789b3cb7dcc220","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-05-06 01:00:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad74f819bf4dd6c1a0e26d1d2bf2d4ead6ca85fc","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-05-07 01:50:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c84b5f4f84a88fa7130393b3106e12f2da769a46","author":{"_account_id":19098,"name":"lumeihong","email":"lu.meihong@zte.com.cn","username":"lu-meihong"},"date":"2020-05-08 02:51:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"745dfa15607c62a208ea48f7ecefb5878e39a377","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-08 12:20:57.000000000","message":"Patch Set 1: Code-Review+1\n\nhttps://review.opendev.org/#/c/721912/1 needs to merge first, please review","accounts_in_message":[],"_revision_number":1},{"id":"ba67785df14f001c930becd2e3e29dac63fbb5b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 01:51:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdca7b7fe3ae88963da347ec820c2b8c9e183da5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 02:25:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bce7e143fe814874b550cb68d3c502d0 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e4ca060d754495843ffdea71eea622 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e9ca421546a4807a47d739f6ef07d68 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce630d6e4320405099e545c660f81484 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f1cf6908054d9a952c947befd4c466 : SUCCESS in 5m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b1f99384af004ffa96b281a019a643e0 : SUCCESS in 5m 12s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/651ee8c318ea4c1693eb3f2590978d40 : SUCCESS in 32m 01s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/311b1982a1704ff4853e5f01d5507099 : SUCCESS in 31m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9b67d5e317eb0d52231128ea92846fa6ec977f2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 02:25:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f4ba9038bab79c97628d9793cef7574f3fe7a489","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 02:27:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96418802d0b14ce0b9c1806891a62534 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b282249460ac2d78ed5bf576d4796024fe6665f","revisions":{"3b282249460ac2d78ed5bf576d4796024fe6665f":{"kind":"REWORK","_number":1,"created":"2020-04-22 13:09:56.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/15/721915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/721915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/721915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/721915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/721915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/721915/1"}}},"commit":{"parents":[{"commit":"427deee2222bbe1a12d66cab25ac2f050d538482","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/427deee2222bbe1a12d66cab25ac2f050d538482"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 13:09:55.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 13:09:55.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Id34ad306a28f77785abf2c84f16a091658558bff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3b282249460ac2d78ed5bf576d4796024fe6665f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3b282249460ac2d78ed5bf576d4796024fe6665f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
