)]}'
{"id":"openstack%2Ffreezer~722361","triplet_id":"openstack%2Ffreezer~master~I16abcb7f3c828b454495f3dc94b904ca91943aa7","project":"openstack/freezer","branch":"master","topic":"bug/1874421","attention_set":{},"removed_from_attention_set":{"31885":{"account":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"last_update":"2024-11-12 06:01:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16abcb7f3c828b454495f3dc94b904ca91943aa7","subject":"fixed the error when mounting a lvm snapshot in which filesystem inconsistencies exist","status":"MERGED","created":"2020-04-23 15:29:29.000000000","updated":"2024-11-12 06:03:34.000000000","submitted":"2024-11-12 06:01:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722361-bug/1874421","meta_rev_id":"89096f1305dd76167aaa124cfc9ec0aba1ff534d","_number":722361,"virtual_id_number":722361,"owner":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-12 06:01:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2024-11-12 05:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2024-11-12 05:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 15:29:57.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-04-23 16:17:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-26 01:47:49.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2022-05-08 09:54:53.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REMOVED"},{"updated":"2024-11-12 05:50:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"d9e0540307368fe460cd7f7f8b5172f100093b83","author":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"date":"2020-04-23 15:29:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68c241657bbfa82d7fd97a588edaf9b14ca9118b","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-04-23 15:29:57.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"44a4805dbde4993d7aab5585fd9145a8533d396f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:17:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/defacfaed1634eb5bcd22d156066ca14 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d31ec1c7218c4c38b6a9c69b3f4c8b91 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9afa3c5016ab497b90b0fb508993c693 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/686017b49edb4041b936ab5bb1897501 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/624a1ce8e3d249ee8e7417956304faca : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44816f309204b389f5a2e33fe091dc3 : SUCCESS in 5m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/396bdb5b97c7498d853469575d2c6109 : SUCCESS in 5m 37s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/55fe40fda2cd49a0999c8f282e1f1b0a : SUCCESS in 34m 48s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/aa9d6ec59b034b3c92a405b275bee4c5 : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"57ee2e8e7e02557093f07ed6b3ed4085979c26dd","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-04-26 01:47:49.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease add unit test cases for the commit.\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"1fa858621daf3bb031b5ce36d893ae3ec9081d48","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2022-05-08 09:54:53.000000000","message":"Removed reviewer Carl caihui with the following votes:\n\n* Code-Review-1 by Carl caihui \u003ccaihui_nj@163.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"a12ccab9f960957d5c2369add1b124c99def832d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 17:21:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_31885\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"}],"_revision_number":2},{"id":"45239906a3d7a3b3aa982db3174e9d812c3ba7b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:29:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/825f568475f7415183bac3fd82681e4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d433f2357164321858c110ea48e12e8 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f47d9cdc91004497b107200c81acd7ff : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75a39cac362c022d8cddbe4786a76d6c1b3aea6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:02:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a20e77ee12947d5802950a64ab2ca1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79f84c6f36d43c29849cbc32c453448 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee741cff5b94036b3cf3dcb72e48605 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4703c0ec71d4fc2840ab27398469a05 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7678075c3014b859aa525a04fb0457d : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2137c14f1cb449a4ac49e9d3dac80bbd : SUCCESS in 3m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3bac3b59302e43b5a1d9d830ec7d810f : SUCCESS in 4m 43s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/df4896e509ab4806a89b900bcb670254 : FAILURE in 37m 20s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/988ea37cd7ed4bd5a153f5a285aca369 : FAILURE in 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c038d9aaa796b944efca1a5efba6b1c127ead070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-30 15:30:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_31885\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"}],"_revision_number":3},{"id":"63c46a8d00813cf880046e8dccd2d3f208973ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 15:36:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed3dd1869b9e46fba566a50fe5708b07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd3e49d88f849748698800a934e2504 : SUCCESS in 3m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8808cdc711da45ce98b2ae41e0859f85 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"e2d0576b3428c8c727c806536deaf1bf3054f696","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-12 05:50:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cd28e01bac15342886599bdc7704fa5bf5f7c1d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 05:50:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"15fa2b035d1eb97ab4f1f107059e996a9b631a0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 06:01:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d2206e4ff94413eb2a83d985f694930\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f666dc206a24429eba2f21d5a4716233 : SUCCESS in 3m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb9f677bc99f41d399a9c2da72bc5ad3 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"f46c55687d1d0d11f685124bedcef8de0958bc95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 06:01:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"89096f1305dd76167aaa124cfc9ec0aba1ff534d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 06:03:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a8b6253e5e44768f92795eae01671c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a6a362123f44cd94997601844f6e10 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8e55e2a3f4dab92d478b452db942f1d0f85336aa","revisions":{"1ac95b06db7ef6783c36e60e6d9e99cff554b3f5":{"kind":"REWORK","_number":1,"created":"2020-04-23 15:29:29.000000000","uploader":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"ref":"refs/changes/61/722361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/61/722361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/61/722361/1"}}},"commit":{"parents":[{"commit":"851a17173b0e6f7f8d6d188406897b09c8b6b50e","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/851a17173b0e6f7f8d6d188406897b09c8b6b50e"}]}],"author":{"name":"Haosheng He","email":"dhs007@qq.com","date":"2020-04-23 15:04:33.000000000","tz":480},"committer":{"name":"Haosheng He","email":"dhs007@qq.com","date":"2020-04-23 15:12:13.000000000","tz":480},"subject":"fixed the error when mounting a lvm snapshot in which filesystem inconsistencies exist","message":"fixed the error when mounting a lvm snapshot in which filesystem inconsistencies exist\n\nwhen we create a lvm snapshot with mode mysql, as no sync commands will be performed, so probably the filesystem of the lvm snapshot is inconsistant. We need to use mount options ro and norecovery to omit inconsistency errors in filesystem.\n\nChange-Id: I16abcb7f3c828b454495f3dc94b904ca91943aa7\nCloses-Bug:1874421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/1ac95b06db7ef6783c36e60e6d9e99cff554b3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/1ac95b06db7ef6783c36e60e6d9e99cff554b3f5"}]},"branch":"refs/heads/master"},"3333855de278f4465a8fe5d6d6d4c8fac629207b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-02 17:21:28.000000000","uploader":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/722361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/61/722361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/61/722361/2"}}},"commit":{"parents":[{"commit":"5effc1382833ad111249bcd279b11fbe95e10a6b","subject":"Merge \"Fix freezer doc error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5effc1382833ad111249bcd279b11fbe95e10a6b"}]}],"author":{"name":"Haosheng He","email":"dhs007@qq.com","date":"2020-04-23 15:04:33.000000000","tz":480},"committer":{"name":"Haus He","email":"he.haosheng@datatom.com","date":"2024-05-02 17:21:28.000000000","tz":0},"subject":"fixed the error when mounting a lvm snapshot in which filesystem inconsistencies exist","message":"fixed the error when mounting a lvm snapshot in which filesystem inconsistencies exist\n\nwhen we create a lvm snapshot with mode mysql, as no sync commands will be performed, so probably the filesystem of the lvm snapshot is inconsistant. We need to use mount options ro and norecovery to omit inconsistency errors in filesystem.\n\nChange-Id: I16abcb7f3c828b454495f3dc94b904ca91943aa7\nCloses-Bug:1874421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3333855de278f4465a8fe5d6d6d4c8fac629207b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3333855de278f4465a8fe5d6d6d4c8fac629207b"}]},"branch":"refs/heads/master","description":"Rebase"},"8e55e2a3f4dab92d478b452db942f1d0f85336aa":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-30 15:30:55.000000000","uploader":{"_account_id":31885,"name":"Haus He","email":"he.haosheng@datatom.com","username":"dhscq"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/722361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/61/722361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/722361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/61/722361/3"}}},"commit":{"parents":[{"commit":"735302b27dc520b754a87644a35c23a767ed7784","subject":"Remove unsupported distros and disable failing checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/735302b27dc520b754a87644a35c23a767ed7784"}]}],"author":{"name":"Haosheng He","email":"dhs007@qq.com","date":"2020-04-23 15:04:33.000000000","tz":480},"committer":{"name":"Haus He","email":"he.haosheng@datatom.com","date":"2024-07-30 15:30:55.000000000","tz":0},"subject":"fixed the error when mounting a lvm snapshot in which filesystem inconsistencies exist","message":"fixed the error when mounting a lvm snapshot in which filesystem inconsistencies exist\n\nwhen we create a lvm snapshot with mode mysql, as no sync commands will be performed, so probably the filesystem of the lvm snapshot is inconsistant. We need to use mount options ro and norecovery to omit inconsistency errors in filesystem.\n\nChange-Id: I16abcb7f3c828b454495f3dc94b904ca91943aa7\nCloses-Bug:1874421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8e55e2a3f4dab92d478b452db942f1d0f85336aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8e55e2a3f4dab92d478b452db942f1d0f85336aa"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
