)]}'
{"id":"openstack%2Ffreezer~737710","triplet_id":"openstack%2Ffreezer~master~I02208d8bc9905b1bc6de0dbccf63107c307ba05b","project":"openstack/freezer","branch":"master","topic":"test_ci","hashtags":[],"change_id":"I02208d8bc9905b1bc6de0dbccf63107c307ba05b","subject":"Update CREDITS list","status":"MERGED","created":"2020-06-24 08:35:00.000000000","updated":"2020-06-24 09:27:38.000000000","submitted":"2020-06-24 09:24:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737710-1592990687664-384920b2","meta_rev_id":"e15f8de2dc520ff582dc08ed4f8b51309553601a","_number":737710,"virtual_id_number":737710,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 09:24:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-24 09:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-24 09:10:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 09:24:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd1be9e873a78adcdb2c45f64ae8d3af02ec8578","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-24 08:35:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a73355d4c2b0f49acddd88d4a78d6acfa4a02d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 08:46:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f547f70ab3cd41f5a784b89f5015a290 : FAILURE in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18a25db037244f691a0b86ae9ee873d : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a51620f75745dd7b0ca917ba968e4555dc882f1c","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-24 08:54:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e50c1536e5a02f94c9ad801d085586873352951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:05:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc58ddebfed4816be9d53e92f61ad22 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26bfbff1bb14479e9419ad44db55b895 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8a01ca5727e664fea62478b9ae03f65f4dd69b67","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-24 09:10:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0fbc2a5bde5e0c7fae00db0607594f2ae5601848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:11:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"488d651c4193b17bd0f37a5db85fbcff93659ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:24:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c898c5e1dc94a299c8b1351b4119a5b : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c39693143e44df38bd24e7e7c1d83cc : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9d13c86e3f24e65c96477ac7cfdf652e3b0a1fb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:24:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fd1b1bdba774ce7cb3c976416da644498fd15fc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:27:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4aff5a198448568a07c3d1f33842c1 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"648ca1e3ba93ef48edd7b192422b93beadb1de2b","revisions":{"f2e38bcd7a7267e763e067a0959df148deff4d93":{"kind":"REWORK","_number":1,"created":"2020-06-24 08:35:00.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/10/737710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/10/737710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/10/737710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/10/737710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/10/737710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/10/737710/1"}}},"commit":{"parents":[{"commit":"a4e758eb34b4ec50f5e3aaa86e3cf7b6583935ab","subject":"Remove packages version for python2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a4e758eb34b4ec50f5e3aaa86e3cf7b6583935ab"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-24 08:31:08.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-24 08:31:08.000000000","tz":-420},"subject":"Update CREDITS list","message":"Update CREDITS list\n\nZTE TECS team  maintains the Freezer project in recent releases.\n\nUpdate CREDITS list.\n\nChange-Id: I02208d8bc9905b1bc6de0dbccf63107c307ba05b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f2e38bcd7a7267e763e067a0959df148deff4d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f2e38bcd7a7267e763e067a0959df148deff4d93"}]},"branch":"refs/heads/master"},"648ca1e3ba93ef48edd7b192422b93beadb1de2b":{"kind":"REWORK","_number":2,"created":"2020-06-24 08:54:40.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/10/737710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/10/737710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/10/737710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/10/737710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/10/737710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/10/737710/2"}}},"commit":{"parents":[{"commit":"a4e758eb34b4ec50f5e3aaa86e3cf7b6583935ab","subject":"Remove packages version for python2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a4e758eb34b4ec50f5e3aaa86e3cf7b6583935ab"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-24 08:31:08.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-24 08:54:30.000000000","tz":-420},"subject":"Update CREDITS list","message":"Update CREDITS list\n\nZTE TECS team  maintains the Freezer project in recent releases.\n\nUpdate CREDITS list.\n\nChange-Id: I02208d8bc9905b1bc6de0dbccf63107c307ba05b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/648ca1e3ba93ef48edd7b192422b93beadb1de2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/648ca1e3ba93ef48edd7b192422b93beadb1de2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
