)]}'
{"id":"openstack%2Ffreezer~737746","triplet_id":"openstack%2Ffreezer~master~I390db6ccb6bf17e6d641ad88e3dd6273750e3c4d","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"I390db6ccb6bf17e6d641ad88e3dd6273750e3c4d","subject":"Update tox env  param","status":"MERGED","created":"2020-06-24 11:02:50.000000000","updated":"2020-06-28 00:23:42.000000000","submitted":"2020-06-28 00:22:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737746-1593303740721-9fea2b71","meta_rev_id":"f9f512baba9cc40c7c818e00e3759486fb986b36","_number":737746,"virtual_id_number":737746,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-28 00:22:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-27 23:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-27 23:53:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-28 00:22:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90991e2a210e9fb5479a8b7160e5d51ffabbcf7d","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-24 11:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1822e54c78228b3ef6ac4c620efccedc67c902f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 12:08:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6124b7b06f4478d83d471b006edb504 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025222114946427fa98eab129a4999a4 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a7eb6bef36949b38ff98b9d93eab756 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e334fed41224057af1857c1375ef16b : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae1b49447e04fab940c4fcc5f313ead : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/153ac9a6037c4147ba95cb93cf6125b5 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34c52aeedf604f48aa3d6bde4b216413 : SUCCESS in 5m 44s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/9da80a7390744975aade00485e13fc1c : SUCCESS in 43m 42s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/dd1e52a817ce41d583d4d316f45c4afe : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":1},{"id":"336c16797d61f47ead84ed60ef852169a2aff1c1","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-27 23:53:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"261b870bc1f4158bc588125659a9a207519493f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 23:54:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50fc990a7c7387dbbaa43f0fd68596c2592142e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:22:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd73a5aba95b4e15bda1651c460879d7 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16aaa31ed6a42e9b7edcbcb863b64b6 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6e06e16ff724779b3b7705a9047ae3c : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9574d24da805447bb65d5daeac8d512e : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26934ba95ec411694a6ec5aadf1723a : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec27861a1c474606bd4524524f355b51 : SUCCESS in 3m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0d2989e48c114f379137febf6cecd748 : SUCCESS in 4m 52s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/5f54ae43cd2a425381a91181db7bbbd6 : SUCCESS in 21m 51s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/b053f860197c47d4b0bf6c1e7401aa47 : SUCCESS in 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"720d263a3fe77873c06936b4dabb1dae88d9d340","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:22:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f9f512baba9cc40c7c818e00e3759486fb986b36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 00:23:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0c2c75dcbd4a1185637f74f2cb6c04 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a872155a1a024df899df877fadda7696 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3aea14c6860d2fad96fe0884cd3a81128a6da06","revisions":{"a3aea14c6860d2fad96fe0884cd3a81128a6da06":{"kind":"REWORK","_number":1,"created":"2020-06-24 11:02:50.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/46/737746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/46/737746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/46/737746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/46/737746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/46/737746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/46/737746/1"}}},"commit":{"parents":[{"commit":"85111f528bd73e44ed7d22985ceadf96e9100c45","subject":"Update \"if...if...\" to \"if...elif...else...\" in BackupJob._validata()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/85111f528bd73e44ed7d22985ceadf96e9100c45"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-24 10:56:52.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-24 10:56:52.000000000","tz":-420},"subject":"Update tox env  param","message":"Update tox env  param\n\nThe \"tenant\" is replaced by \"project\"  in new version.\n\nChange-Id: I390db6ccb6bf17e6d641ad88e3dd6273750e3c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a3aea14c6860d2fad96fe0884cd3a81128a6da06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a3aea14c6860d2fad96fe0884cd3a81128a6da06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
