)]}'
{"id":"openstack%2Ffreezer~738414","triplet_id":"openstack%2Ffreezer~master~I06cda902c6d81b415a7f0baf083a9c14b3dbf89d","project":"openstack/freezer","branch":"master","topic":"ut_test","hashtags":[],"change_id":"I06cda902c6d81b415a7f0baf083a9c14b3dbf89d","subject":"Test cases  can run in  python3.x runtime Test cases  of test_rsync can run in  python3.x runtime.","status":"MERGED","created":"2020-06-29 10:42:52.000000000","updated":"2020-06-29 12:16:56.000000000","submitted":"2020-06-29 12:14:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738414-1593432893246-9fea2b71","meta_rev_id":"65dfcc8b03373c9fb5de6c8810ea893a6587e4b1","_number":738414,"virtual_id_number":738414,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-29 12:14:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-29 11:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-29 11:36:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 12:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8e2ba400a80f8cd8751007b5c5e63b28301800d","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-29 10:42:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52196a0d0c9f53711913bf036562a75b316938ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 11:18:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f05611276b1a4060ad2a32dd21b1661a : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651d4345951d42b482c1a0d3f8f77ee3 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42bf4ceb48a2470c8b12618654247bf0 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/561cf6c743c24d4891f187b80d10a82c : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8fd24b5b6f46ec96fbe2d043446916 : SUCCESS in 5m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e27da8e4a6d44a2ea5821e25ea2e9f7b : SUCCESS in 6m 51s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/253928c0d57446a9b0a1dfa5e48b2503 : SUCCESS in 34m 33s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/ce58016f68cc4d72a48c4a2c809009e6 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3d00a481dd01dceebe78a645e4773140ad1d5db4","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-29 11:36:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dba893eb5bb10e0a722a1748242ea401657261b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 11:36:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7a95193a2ce37f64fbe55d23a63f805313d6466","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 12:14:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fda394c7bc44e9d8a27631e1944890a : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b446d27dc94bb784124e5f42dc9a47 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2db080ae34034836b43dcf06b8d5284c : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ef76b429b1e450a9a9f6bf3e12df3c0 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06cac519445f4b40b2198ebd819f2483 : SUCCESS in 6m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d1035c955294d0e9106ec269c823fba : SUCCESS in 5m 50s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/5b33ac397a84425294ddcaec74086ec2 : SUCCESS in 36m 58s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/085c767ca45944c09bd7870c481b84b7 : SUCCESS in 33m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d216e62f39297706441a5bb6f2e50b4a3f8a0654","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 12:14:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bede2c9162cdb945c1f80c520312d2bdec31e991","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 12:16:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f3d7bd65fe4fc7ab20b9e86072fb02 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdda55acfbf6e54dc3dc05e008e09c81a778b6f9","revisions":{"fdda55acfbf6e54dc3dc05e008e09c81a778b6f9":{"kind":"REWORK","_number":1,"created":"2020-06-29 10:42:52.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/14/738414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/14/738414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/14/738414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/14/738414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/14/738414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/14/738414/1"}}},"commit":{"parents":[{"commit":"564675a7f58a43b8268dc1448e5c5ec7f7e683d2","subject":"Some test cases in test_ftp support python3.X","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/564675a7f58a43b8268dc1448e5c5ec7f7e683d2"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-29 10:38:10.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-29 10:42:33.000000000","tz":-420},"subject":"Test cases  can run in  python3.x runtime Test cases  of test_rsync can run in  python3.x runtime.","message":"Test cases  can run in  python3.x runtime\nTest cases  of test_rsync can run in  python3.x runtime.\n\nChange-Id: I06cda902c6d81b415a7f0baf083a9c14b3dbf89d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fdda55acfbf6e54dc3dc05e008e09c81a778b6f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fdda55acfbf6e54dc3dc05e008e09c81a778b6f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
