)]}'
{"id":"openstack%2Ffreezer~738424","triplet_id":"openstack%2Ffreezer~master~Ia03aef1f5e51403f5ef1fdba65d0fa231b8889fd","project":"openstack/freezer","branch":"master","topic":"ut_test","hashtags":[],"change_id":"Ia03aef1f5e51403f5ef1fdba65d0fa231b8889fd","subject":"Fix a bug to support python3.x","status":"MERGED","created":"2020-06-29 11:36:04.000000000","updated":"2020-06-30 04:27:28.000000000","submitted":"2020-06-30 04:26:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738424-1593491173370-3d5ab54e","meta_rev_id":"a2a0d730e6126b13a5ed49c3cc04084a651782e8","_number":738424,"virtual_id_number":738424,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 04:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-30 03:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-30 03:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 04:26:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac5ad7f67e604d1f28b4311182e07f454299a22d","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-29 11:36:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"873bbfd18bff57b5481d0182fec3d7d56479d23b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 12:16:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b40189629c4f4eab994e47fefd103747 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c199498bb32d4d96b1f5a67d926f5c9f : FAILURE in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c7de21c3df47f88fed9d55306c19e9 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b749adbdac740cc91620d76279b1da1 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cdaf3a65f264ec5a6a139a49483ffc8 : SUCCESS in 5m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cfe1e71eacda4be9b5434f4dab524656 : SUCCESS in 4m 59s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/9182a683a0c64b4dbb0eb941c57172bf : SUCCESS in 34m 23s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/43bb13f0f9454aa88085555244cb8e4c : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":1},{"id":"57933c224368c4a47e1fbdd926d905899f5db335","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-30 01:41:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64401c0067a46bc89040e9ca2fe325948f6b4d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 02:22:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f5f1459280c4f51b3e2fda7848591bd : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860142165d4846d1a1d21e3ce3535e65 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/962bc53522c54244baa0918375a3289a : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13181c02fab5443da166a47678b64d89 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58cad011be57480e8f9479c4bbc4e2a6 : SUCCESS in 4m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/66fa7595c60b40018e167e2395b0fe69 : SUCCESS in 6m 27s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/820745654fe84d10994422a1cebb0dfb : SUCCESS in 33m 47s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/fd764893ec7942fb8a75d190e9ea9734 : SUCCESS in 39m 36s","accounts_in_message":[],"_revision_number":2},{"id":"dfee4201491d5d2547b2115e73dd6d2127864cc2","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-30 03:55:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4b9b1ab461cf3bcc7b0677f84e830479add1e45e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 03:55:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07024420441f5a1e6daf5b898bc354f4b751ce9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 04:26:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01278c6fbcb2467cb4abec80211facb0 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c95fca00aa624b059178ba7d8384a563 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94393b204af34b8c99e2fa73334c67d7 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4593fa084b4151a6468bb2ecf5b350 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497bdd80480048a790af1db9fa7ba4d2 : SUCCESS in 5m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81bdbde0e6f14c73aff05b24021c9811 : SUCCESS in 6m 34s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/0c083a88b5ac402a8c5c3e5a77e7a9fd : SUCCESS in 29m 47s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/8229df0e804f4404b0e2fcdb0b013fc5 : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ea1cb45682f1bb5ef83f78b34a1c56d2472631c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 04:26:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"afbc6e82b4ccb39c0beaeb6f121351ad2c78345f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 04:27:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f73b5871aff45f18ba6f6d71f1225bb : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2a46346d3ccb2ead8c9b09d3d33d72f86311360","revisions":{"c202a447f238a38915f9fb119951ea47931c1a39":{"kind":"REWORK","_number":1,"created":"2020-06-29 11:36:04.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/24/738424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/24/738424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/738424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/738424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/738424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/24/738424/1"}}},"commit":{"parents":[{"commit":"fdda55acfbf6e54dc3dc05e008e09c81a778b6f9","subject":"Test cases  can run in  python3.x runtime Test cases  of test_rsync can run in  python3.x runtime.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fdda55acfbf6e54dc3dc05e008e09c81a778b6f9"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-29 11:31:15.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-29 11:31:15.000000000","tz":-420},"subject":"Fix a bug to support python3.x","message":"Fix a bug to support python3.x\n\nThe default type of the string is str in python3.x,\nif want use the type of byte ,should be delcare b\u0027xxxx\u0027 .\n\nChange-Id: Ia03aef1f5e51403f5ef1fdba65d0fa231b8889fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c202a447f238a38915f9fb119951ea47931c1a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c202a447f238a38915f9fb119951ea47931c1a39"}]},"branch":"refs/heads/master"},"d2a46346d3ccb2ead8c9b09d3d33d72f86311360":{"kind":"REWORK","_number":2,"created":"2020-06-30 01:41:09.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/24/738424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/24/738424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/738424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/738424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/24/738424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/24/738424/2"}}},"commit":{"parents":[{"commit":"fdda55acfbf6e54dc3dc05e008e09c81a778b6f9","subject":"Test cases  can run in  python3.x runtime Test cases  of test_rsync can run in  python3.x runtime.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fdda55acfbf6e54dc3dc05e008e09c81a778b6f9"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-29 11:31:15.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-30 01:40:53.000000000","tz":-420},"subject":"Fix a bug to support python3.x","message":"Fix a bug to support python3.x\n\nThe default type of the string is str in python3.x,\nif want use the type of byte ,should be delcare b\u0027xxxx\u0027 .\n\nChange-Id: Ia03aef1f5e51403f5ef1fdba65d0fa231b8889fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2a46346d3ccb2ead8c9b09d3d33d72f86311360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2a46346d3ccb2ead8c9b09d3d33d72f86311360"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
