)]}'
{"id":"openstack%2Ffreezer~738569","triplet_id":"openstack%2Ffreezer~master~I558bbfecb4a9670a10b1fc649821d88619b7917c","project":"openstack/freezer","branch":"master","topic":"ut_test","hashtags":[],"change_id":"I558bbfecb4a9670a10b1fc649821d88619b7917c","subject":"Fix bug \"AttributeError: \u0027generator\u0027 object has no attribute \u0027next\u0027\"","status":"MERGED","created":"2020-06-30 09:11:42.000000000","updated":"2020-06-30 10:51:03.000000000","submitted":"2020-06-30 10:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738569-1593514151062-886363f1","meta_rev_id":"be96bc315124ed5ee0c6fc70abf0a99a24562f9b","_number":738569,"virtual_id_number":738569,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 10:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-30 10:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-30 10:05:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 10:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2719d821d3635c79a1e8d00720ad22af8048510","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-30 09:11:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bab6951932885d539b8ef7c9cd0bacf9bb3470d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 09:54:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dac83041db62458e90b986f5c1914454 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200073e80cfb463da61129549b48e21a : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/566b086c6a9043489d64dc997417a6ec : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7afb85325234e26a980da83e2c58c08 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ee642171d84e1998872d78659e3710 : SUCCESS in 6m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7245b1e996d04d029e5624078c554bf1 : SUCCESS in 7m 07s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/ec6069366d614cf0b53c9be9a107887d : SUCCESS in 36m 14s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/b1bcdac481da4b5f8081eddca60cb451 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"05a109f814e883f3f4ce95cb75d90412cecce472","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-30 10:05:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2468ba866566fed2d61972d2887d3298dc981575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:05:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1256a56191cdc5e553246d71e12e5379b655cdfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:49:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78de65e3e2184f89b26252bb88e3b0bd : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd749289fea244bcb1e68d1b6e5354ee : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c0e9b285ca44713928dc88caed7d118 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5e1d4f043964a3ca14868cfba0e1bf1 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da45865be51b4fbf84b6bec4e0a4cccd : SUCCESS in 4m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2954b18d8f184c38b0d8b2ee7e76fdcd : SUCCESS in 6m 30s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/b8ced11429a845ba88bad5719dc5a5cd : SUCCESS in 36m 05s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/d70ad49ba1dc4126bce49bd98c21139a : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e48452fe73ae4a892ed64e52aaad37e549a9aecc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:49:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b283cab673ea48cee93d48b47af3e9a758747bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:51:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4faae06bd88d43579963d4c509c3861d : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19aae78daf61662ca96206d524173030c421fc50","revisions":{"19aae78daf61662ca96206d524173030c421fc50":{"kind":"REWORK","_number":1,"created":"2020-06-30 09:11:42.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/69/738569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/69/738569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/738569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/738569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/69/738569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/69/738569/1"}}},"commit":{"parents":[{"commit":"7236b3225e1b5b07e2fb8819d87ed254928c1f4a","subject":"Fix bug in python3.x: \u0027bytes\u0027 object has no attribute \u0027format\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/7236b3225e1b5b07e2fb8819d87ed254928c1f4a"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-30 08:58:27.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-30 08:58:27.000000000","tz":-420},"subject":"Fix bug \"AttributeError: \u0027generator\u0027 object has no attribute \u0027next\u0027\"","message":"Fix bug \"AttributeError: \u0027generator\u0027 object has no attribute \u0027next\u0027\"\n\n1. the bug:\n   Captured traceback:\n   ~~~~~~~~~~~~~~~~~~~\n\tTraceback (most recent call last):\n\n\tFile \"/opt/stack/freezer/freezer/tests/unit/scheduler/test_utils.py\", line 53, in test_find_config_files_path\n\tret \u003d utils.find_config_files(temp_path)\n\n\tFile \"/opt/stack/freezer/freezer/scheduler/utils.py\", line 53, in find_config_files\n\tfor fname in os.walk(expanded_path).next()[2]:\n\n\tAttributeError: \u0027generator\u0027 object has no attribute \u0027next\u0027\n\n2. fix:\nTo get iterobject next element  is  changed to next(iterobject,defalt) in python3.x.\n\nChange-Id: I558bbfecb4a9670a10b1fc649821d88619b7917c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/19aae78daf61662ca96206d524173030c421fc50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/19aae78daf61662ca96206d524173030c421fc50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
