)]}'
{"id":"openstack%2Ffreezer~738588","triplet_id":"openstack%2Ffreezer~master~Ic912cc86d9a5afbdbcd63f27e1f91d2140f9b9ba","project":"openstack/freezer","branch":"master","topic":"bug_fix","hashtags":[],"change_id":"Ic912cc86d9a5afbdbcd63f27e1f91d2140f9b9ba","subject":"Fix Type error","status":"MERGED","created":"2020-06-30 11:40:16.000000000","updated":"2020-07-01 00:41:38.000000000","submitted":"2020-07-01 00:40:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738588-1593564014259-fcaafab0","meta_rev_id":"66b95c915df335b992e8db9b287a1e4c6e57b7cd","_number":738588,"virtual_id_number":738588,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 00:40:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-30 23:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-30 23:59:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-01 00:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0043b731e4e98c562f4cbf950394db8790dfb24d","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-30 11:40:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"101b0f2b456c2a1bca95f6f71086123c92c80333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 12:23:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19a5183b75d54963a65214214a6d5c54 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517d244514db406faf16658b208537ac : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccc83d21f54847ce8e3befbd7221f13c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6db21a010c4b4899892b6762c2fdaed7 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031565d5591244cd8b543abae7e804ad : SUCCESS in 4m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d36453b261cc46e9ad40574c60075e17 : SUCCESS in 4m 51s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/082074681a6f4ae587dd220cc5d8cae6 : SUCCESS in 36m 39s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/e9207bbb9ef6408ab6f35235b9cb9f78 : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":1},{"id":"be22a35a8a8bf98952581fbae7fee5fad1298e29","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-06-30 23:59:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c5cb3056ed4c93ede3580be5b540edc64697708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 23:59:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2f2a667e5663fa527d9c3dfb83b3817c807dd58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 00:40:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65d058bbfcfe4bc1866597bb8dd62ad9 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c15668e8139a4d9c88bcc1692fc9a2f9 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18b619da69d845b2834d52001b30c989 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f608bd19714f04bf454e63b2c77cc6 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f229a75006141b1b6c59e2dd83fe825 : SUCCESS in 5m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e832237845454aa5afafb736042417e3 : SUCCESS in 4m 55s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/db6586279819490daf53dee1dba9bfca : SUCCESS in 35m 09s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/88592f1850604819b97ccaf7eddf1838 : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7e048a89f958e0ef25fa148d36387935d1608468","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 00:40:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3ec31aa87e8df7615267809e4232ff4eace480f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 00:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf46517f9014c529f327979d821f51c : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ae26c12fb03a30133c62c9b79b4552cb32f0717","revisions":{"3ae26c12fb03a30133c62c9b79b4552cb32f0717":{"kind":"REWORK","_number":1,"created":"2020-06-30 11:40:16.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/88/738588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/88/738588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/738588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/738588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/738588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/88/738588/1"}}},"commit":{"parents":[{"commit":"19aae78daf61662ca96206d524173030c421fc50","subject":"Fix bug \"AttributeError: \u0027generator\u0027 object has no attribute \u0027next\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/19aae78daf61662ca96206d524173030c421fc50"}]}],"author":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-30 11:29:32.000000000","tz":-420},"committer":{"name":"Caihui","email":"cai.hui@zte.com.cn","date":"2020-06-30 11:36:59.000000000","tz":-420},"subject":"Fix Type error","message":"Fix Type error\n\nThe \"file_head\" is a \u0027bytes-like object ,not \u0027str\u0027.\n\nUpdate file_head.split() to file_head.decode(\u0027utf-8\u0027).split()\n\nChange-Id: Ic912cc86d9a5afbdbcd63f27e1f91d2140f9b9ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3ae26c12fb03a30133c62c9b79b4552cb32f0717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/3ae26c12fb03a30133c62c9b79b4552cb32f0717"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
