)]}'
{"id":"openstack%2Ffreezer~741077","triplet_id":"openstack%2Ffreezer~master~Iffbd3ea1195851b3353edff65f9c1a91da8f09cc","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Iffbd3ea1195851b3353edff65f9c1a91da8f09cc","subject":"Fix the bug","status":"ABANDONED","created":"2020-07-15 03:43:00.000000000","updated":"2020-07-15 09:04:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df80d48b252e2b3e61a983c959fa45e247c3b951","_number":741077,"virtual_id_number":741077,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-15 04:24:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-15 07:30:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 04:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-15 07:30:37.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"eea48e244486d86fe0e2ee0c35b843a4b297e06d","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-07-15 03:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74d80ad03ce6086aee20787b230c53412c2f217d","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-07-15 03:44:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2f2faaeae8b68c5cf391ffc16f8cb3e1f957537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 04:24:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54ce015618744d4fb759b916fd916018 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf8809103274b65aea83315154aef85 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f31bc709f0e94fbcb9c01950337765b2 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3e976dae92a40648926084a2fdcc4aa : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f8ddbd44534c3baabc352590b5e65a : SUCCESS in 5m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2177fe9a43a54a759ae5fe714d9e4ea4 : SUCCESS in 5m 23s\n- freezer-agent-tempest https://zuul.opendev.org/t/openstack/build/7183d70490834c24a9364908553335da : FAILURE in 35m 36s\n- freezer-scheduler-tempest https://zuul.opendev.org/t/openstack/build/f27a57a2b55a4a7596e829d00445c901 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d6a3be8778573914f75f608d902b648513596d40","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-07-15 07:30:37.000000000","message":"Patch Set 2: Code-Review+1\n\ncommit message should include bug link. like: Closes-Bug: #123456","accounts_in_message":[],"_revision_number":2},{"id":"df80d48b252e2b3e61a983c959fa45e247c3b951","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-07-15 09:04:37.000000000","message":"Abandoned\n\nThe fixed code is error.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2619677fac50e8ba906daeef1fb160b47d827f4","revisions":{"ae9b06ab07bc9578e7646c8d820fbfd97263d542":{"kind":"REWORK","_number":1,"created":"2020-07-15 03:43:00.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/77/741077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/77/741077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/77/741077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/77/741077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/77/741077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/77/741077/1"}}},"commit":{"parents":[{"commit":"62bdc1f0f6f66fbfcbbc0d95fca171e8b17d8a74","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/62bdc1f0f6f66fbfcbbc0d95fca171e8b17d8a74"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-07-15 03:24:19.000000000","tz":-420},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-07-15 03:41:44.000000000","tz":-420},"subject":"Fix the bug","message":"Fix the bug\n\nFrom vivek1516:\n\nI am getting below error while using Freezer-client command in openstack\nenvironment.\n\nUploading image to storage path:\nfreezer-agent --action backup --mode nova\n--nova-inst-id f13111ef-c827-4f19-a0d5-894c24cf69c9 --storage local\n--container /home/stack/test-bkp --backup-name test-bkp\n--path-to-backup /home/stack/ --no-incremental true\n--engine nova --log-file test1-bkp.log\n\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\n[-] a bytes-like object is required, not \u0027str\u0027:\n TypeError: a bytes-like object is required, not \u0027str\u0027\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\nTraceback (most recent call last):\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\nFile \"/usr/local/lib/python3.6/site-packages/freezer/utils/streaming.py\",\n line 112, in run\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\nsuper(QueuedThread, self).run()\n\nChange-Id: Iffbd3ea1195851b3353edff65f9c1a91da8f09cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ae9b06ab07bc9578e7646c8d820fbfd97263d542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ae9b06ab07bc9578e7646c8d820fbfd97263d542"}]},"branch":"refs/heads/master"},"f2619677fac50e8ba906daeef1fb160b47d827f4":{"kind":"REWORK","_number":2,"created":"2020-07-15 03:44:53.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/77/741077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/77/741077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/77/741077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/77/741077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/77/741077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/77/741077/2"}}},"commit":{"parents":[{"commit":"62bdc1f0f6f66fbfcbbc0d95fca171e8b17d8a74","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/62bdc1f0f6f66fbfcbbc0d95fca171e8b17d8a74"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-07-15 03:24:19.000000000","tz":-420},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-07-15 03:44:34.000000000","tz":-420},"subject":"Fix the bug","message":"Fix the bug\n\nFrom vivek1516:\n\nI am getting below error while using Freezer-client command in openstack\nenvironment.\n\nUploading image to storage path:\nfreezer-agent --action backup --mode nova\n--nova-inst-id f13111ef-c827-4f19-a0d5-894c24cf69c9 --storage local\n--container /home/stack/test-bkp --backup-name test-bkp\n--path-to-backup /home/stack/ --no-incremental true\n--engine nova --log-file test1-bkp.log\n\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\n[-] a bytes-like object is required, not \u0027str\u0027:\n TypeError: a bytes-like object is required, not \u0027str\u0027\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\nTraceback (most recent call last):\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\nFile \"/usr/local/lib/python3.6/site-packages/freezer/utils/streaming.py\",\n line 112, in run\n2020-06-10 14:03:23.371 22320 ERROR freezer.utils.streaming\nsuper(QueuedThread, self).run()\n\nChange-Id: Iffbd3ea1195851b3353edff65f9c1a91da8f09cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f2619677fac50e8ba906daeef1fb160b47d827f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/f2619677fac50e8ba906daeef1fb160b47d827f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
