)]}'
{"id":"openstack%2Ffreezer~752503","triplet_id":"openstack%2Ffreezer~master~Id4637ba69793f4c93c90880db9c62d707597dab6","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Id4637ba69793f4c93c90880db9c62d707597dab6","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-09-17 16:23:09.000000000","updated":"2020-09-30 05:35:49.000000000","submitted":"2020-09-30 05:34:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752503-1601444065428-5fc60a99","meta_rev_id":"054b01615ff5d3d443f8620c69d2916692965521","_number":752503,"virtual_id_number":752503,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 05:34:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2020-09-30 05:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-30 03:31:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2020-09-30 05:00:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 03:31:04.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-09-30 05:00:35.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-09-30 05:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"574a85aaa53fa35952cf532c61c35113293507c4","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-17 16:23:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6519e6eb19feb0d9af7940f8d5338d4530d0acdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 17:04:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/85b170119bd64986b6558838a17f553f : SUCCESS in 2m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9aae76fa24e540e494f2559fd27c58ec : FAILURE in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9113f4e8165d47b4bf99c648536f1518 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eeb9113c5e94ff59cb667420241ef96 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b34e8fd39c4414ab1ce26d3a0901a73 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baff0c47a6d149948e579f0ceb84df80 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e6733357bb644a1a9e7584368eeee11 : SUCCESS in 5m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4d326b7590d4dce930bcea8039ccf70 : SUCCESS in 5m 51s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/1dc049d155124724803ef5e03b2e3bb9 : SUCCESS in 36m 56s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e6d2acf419d74a948ad41d97e0cd7ed5 : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bdf4127031a2d11e9f2400a93bd27facae8a10be","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-09-30 02:44:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d9b150fcc53790399567694323a3acb57e9f4c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 03:17:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6cac98acaae45f3a1db4d9bd9b3ad55 : SUCCESS in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b228fdad31c44bbe90027eb5951b57ba : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e1ee1e43b14a9c8911c51fa004f65c : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e3ec14e90c64036b752dc8ecb9036c4 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6979bda3c2064c9bb494677a78f17402 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8de07b213e14b67b659d68b579a2a0c : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b73c8d49573492c88f7c764a8b1ae1e : SUCCESS in 3m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2c41c7fc05845a394a1d977a2e27817 : SUCCESS in 4m 44s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b3b32e6e3e1d442a8bb8398178cd2f4c : SUCCESS in 31m 56s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/af972538b033443593c6af916eef1f1f : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c797dd1be722c7cae80650086d5dfa782c9b6f4a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-30 03:31:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"34dc6d00dbe14c321dbb5a5a03fc802dda8a37fb","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-09-30 05:00:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd95b89a142e21bbf14216568a9a11d2f3059434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:00:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c1bdf729e43f2e070a39348e42531032479f02a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:34:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5b265f79a349b2d79c1a9e6821179cc161b3df1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:34:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/40a65f72360d40ae8126fd3063c24a1b : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76d0512def85448c80ab5deeb0cc272d : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2ddbaceb29e4035b83b01d35a4897f0 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30723681b5fa4f978f756e1945a480a6 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01870832916049f897c08288e3695fa7 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4426d89fa0e14219851db2e5a8831649 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24e1c33df98d41e0ae04148f99949bf2 : SUCCESS in 3m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/82b9152bc91b4330a81c3835a1745a52 : SUCCESS in 5m 55s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/cd2519267a2b43ebbac9c129c4065da0 : SUCCESS in 31m 34s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/deb757ee81954fe5945477f7167ceb9c : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":2},{"id":"054b01615ff5d3d443f8620c69d2916692965521","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 05:35:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98550f9c58174e1693230499d6c5f8b3 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/043fec63c4c743dd9b133dba12da2a8f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b9ac52b1cce226fa7e2053c3cf3e736a72c14662","revisions":{"fa48602f106c2a88a2abe1831573d017f21d0d3d":{"kind":"REWORK","_number":1,"created":"2020-09-17 16:23:09.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/03/752503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/03/752503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/03/752503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/03/752503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/03/752503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/03/752503/1"}}},"commit":{"parents":[{"commit":"663737694ff9129fae222dd6fe0b7253c29490f3","subject":"zuul: use the native jobs from freezer-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/663737694ff9129fae222dd6fe0b7253c29490f3"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-17 16:22:52.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-17 16:23:07.000000000","tz":480},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: Id4637ba69793f4c93c90880db9c62d707597dab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fa48602f106c2a88a2abe1831573d017f21d0d3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/fa48602f106c2a88a2abe1831573d017f21d0d3d"}]},"branch":"refs/heads/master"},"b9ac52b1cce226fa7e2053c3cf3e736a72c14662":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-30 02:44:30.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/03/752503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/03/752503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/03/752503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/03/752503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/03/752503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/03/752503/2"}}},"commit":{"parents":[{"commit":"d037d12e3e35a9b73ad981bf184686a452dcd1a5","subject":"Merge \"[doc] remove six from requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d037d12e3e35a9b73ad981bf184686a452dcd1a5"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-17 16:22:52.000000000","tz":480},"committer":{"name":"gecong","email":"ge.cong@zte.com.cn","date":"2020-09-30 02:44:30.000000000","tz":0},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: Id4637ba69793f4c93c90880db9c62d707597dab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9ac52b1cce226fa7e2053c3cf3e736a72c14662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9ac52b1cce226fa7e2053c3cf3e736a72c14662"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
