)]}'
{"id":"openstack%2Ffreezer~754802","triplet_id":"openstack%2Ffreezer~master~I21a5bf5c11395d76cb35790c5cc985f711f80565","project":"openstack/freezer","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"I21a5bf5c11395d76cb35790c5cc985f711f80565","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-28 16:25:05.000000000","updated":"2020-09-30 01:41:44.000000000","submitted":"2020-09-30 01:38:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754802-1601429890034-68a60c26","meta_rev_id":"c4091749950517f4aa1c1382d826bbc371969e7e","_number":754802,"virtual_id_number":754802,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 01:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"recommended":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"all":[{"value":1,"date":"2020-09-29 20:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":2,"date":"2020-09-30 00:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":1,"date":"2020-09-30 00:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 20:38:48.000000000","updated_by":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"reviewer":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"state":"REVIEWER"},{"updated":"2020-09-30 00:55:51.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-09-30 01:38:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74a9a1112227283f519247c5d5d8a93863111f73","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-28 16:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3a33ac6dea0bacee220601b100c09024d19fc46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 17:20:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c72a60ec54654f109a29bef56d4bdba5 : FAILURE in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ee8ad0a4564b14b5dfd52e98ba61f9 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eb8d9b37cbc4fa8b1512069cca8c24c : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5108257c0144061bc2a32cbc8152a05 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc058d444bf497da94414424ec60050 : SUCCESS in 4m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8999c6c525244925a4dda3a79b0f5998 : SUCCESS in 5m 03s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/94107dfb22ab48d096fb6f0ac5d0cbc0 : FAILURE in 37m 50s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/429398c2948f41d894c7d8848ba35d89 : FAILURE in 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"dd75e782512da0cdb261fbe58475a7aa71b8d6d9","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2020-09-29 08:11:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4728859e0c0c43e13f58e3324607fb19ceb58ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 09:06:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c574803bc834e0ca7cb6699af73e6b7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a866e20e38304f63ac95e36dc1bf5b3f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/130441d003d84fb9a55703f52db4f6f7 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3149ecb1a73d4b52801cb541c359418d : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e865f7cbe444c69544b7a6edcff4e2 : SUCCESS in 6m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bed50b9fc9e44ad59a9e8fc4bdf44cc7 : SUCCESS in 6m 05s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f4eaf06659454896be3fc2f5912af965 : SUCCESS in 37m 10s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/21ee56e494784648beb7e4714408d689 : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2a147148c363ec191ac57bcc760d1709a5a20709","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2020-09-29 20:38:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ad960640401d660e04b6675544ed064307ecabc","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-09-30 00:55:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e14e626bcbd1a39c98a3fd8b66fa218f5a11e33a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 00:56:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3343a0cb83b53c492cf0cdb1b1475693b0412d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 01:38:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5ab8a0c3a784729b7f3c4bcc9e4b508 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025f49011e464613b24c54cfe3e20b86 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21df3626ae9e40148d97b0831b627bb5 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b3b8827c6c45929fc702e9fcffb66a : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cb461d06e843eaaaf284445038beee : SUCCESS in 7m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fedb3c3914614a419ad6d958f79d8a45 : SUCCESS in 4m 21s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/03d21c14ff8448b9ac75ceffd00f97cc : SUCCESS in 36m 45s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/9056832e90a341958c212e5b304ef616 : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b78b9594baa8843bbddc3028adde42e39cbc5e18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 01:38:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d0b3658fec4f0fd503c2423f7f039dc208fe19f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 01:41:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a82dd34e28455f8d0c8ef939bcdc1c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c093f785b678afb60d72e4c2ebe6374b3c6c02cc","revisions":{"c093f785b678afb60d72e4c2ebe6374b3c6c02cc":{"kind":"REWORK","_number":1,"created":"2020-09-28 16:25:05.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/02/754802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/02/754802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/02/754802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/02/754802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/02/754802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/02/754802/1"}}},"commit":{"parents":[{"commit":"5e5d2ee14031749cf00abe4cf157f25fce9b3cc6","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5e5d2ee14031749cf00abe4cf157f25fce9b3cc6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 16:25:04.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-28 16:25:04.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I21a5bf5c11395d76cb35790c5cc985f711f80565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c093f785b678afb60d72e4c2ebe6374b3c6c02cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c093f785b678afb60d72e4c2ebe6374b3c6c02cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
