)]}'
{"id":"openstack%2Ffreezer~760485","triplet_id":"openstack%2Ffreezer~master~I109c38d754fc8c6fedef9947473d29b7ab122e7d","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"I109c38d754fc8c6fedef9947473d29b7ab122e7d","subject":"update package metadata","status":"MERGED","created":"2020-10-30 03:08:02.000000000","updated":"2020-10-30 06:10:25.000000000","submitted":"2020-10-30 06:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760485-1604038131948-d96e2c5f","meta_rev_id":"cbc09477560a7a4e7d4cc6b3291b2b0c042ef357","_number":760485,"virtual_id_number":760485,"owner":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 06:08:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-30 05:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-30 05:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 06:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07deb5fee05ce9c8b3ab8a6d3f7564c3a0f2d4f6","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-10-30 03:08:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca5ebb23d08fbddd5a0171c961d330900d797662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 04:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/31a501d7437d43a29586c6fa601b0228 : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/567c915fd52149a5899caac4fdbb93ac : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbde193a5ca4767a2c16d77b0bc1ff9 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7508edc35054022a6a03560baa40c2c : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6702b489b94436ca3ed400bf510deab : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f5e682eac84bf8bf2202fca7d91f42 : SUCCESS in 5m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d685f1a3aa624657ba6551f8165cb994 : SUCCESS in 5m 11s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/6d91489364b5455d82b943c4bbd6a76b : SUCCESS in 54m 34s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/80caae764efc48319b7cabb766062d2f : SUCCESS in 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"de67ef1510a185d82fe0442c45ed2a26b424adaf","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-10-30 05:22:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb17e031eafb9423db8d130dece51e21e9d6e82b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 05:22:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"544c9b7dd58e4efda2ccee5cc6837d90948d1e57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 06:08:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/95bf394c544b43fd890b9f6db49d7263 : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c0b439b25f04eb38f29165b2e930500 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c77b1d4fcc4e1686723cf1af50f9d0 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88edf8707f024c31aaca28c09e247a05 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32fd3cda5c0d4f3584adb2b882b20930 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055f89cd725e42749b528a0640692e1f : SUCCESS in 5m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bee2e3eb4e6645bbb3dbcd9c90490dde : SUCCESS in 5m 22s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a09bd90785684c6e9eefa022a5be6621 : SUCCESS in 38m 53s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/3b86e1541e3544e2a6727b88ebe4ddf7 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9d17ead24cedb1908b2453c3aaf96b871003aad9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 06:08:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cbc09477560a7a4e7d4cc6b3291b2b0c042ef357","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 06:10:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49bbb2bfa1a4446386175b973520e592 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb0c79396c56dca02c89549df94919ca92ba81e9","revisions":{"cb0c79396c56dca02c89549df94919ca92ba81e9":{"kind":"REWORK","_number":1,"created":"2020-10-30 03:08:02.000000000","uploader":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"ref":"refs/changes/85/760485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/85/760485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/85/760485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/85/760485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/85/760485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/85/760485/1"}}},"commit":{"parents":[{"commit":"0d1368a7a455cad4b025793d79eaf9922db78c6a","subject":"update project FAQ doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0d1368a7a455cad4b025793d79eaf9922db78c6a"}]}],"author":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-10-30 03:07:17.000000000","tz":-420},"committer":{"name":"caihui","email":"cai.hui@zte.com.cn","date":"2020-10-30 03:07:17.000000000","tz":-420},"subject":"update package metadata","message":"update package metadata\n\nChange-Id: I109c38d754fc8c6fedef9947473d29b7ab122e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cb0c79396c56dca02c89549df94919ca92ba81e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cb0c79396c56dca02c89549df94919ca92ba81e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
