)]}'
{"id":"openstack%2Ffreezer~761488","triplet_id":"openstack%2Ffreezer~master~Ia178c8fb54ec88e40af9fd63d00ea3d12b44c7ab","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Ia178c8fb54ec88e40af9fd63d00ea3d12b44c7ab","subject":"The docs  requirements migrated to doc/requirements.txt we need not install things from requirements.txt.","status":"MERGED","created":"2020-11-05 02:29:04.000000000","updated":"2020-11-11 07:39:46.000000000","submitted":"2020-11-11 07:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761488-1605080298276-96dc3111","meta_rev_id":"99ea8d3452d699f89d3c192d1f127ecd6899a46c","_number":761488,"virtual_id_number":761488,"owner":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-11 07:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2020-11-11 06:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-10 07:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2020-11-11 06:34:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 07:17:09.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2020-11-11 06:34:09.000000000","updated_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2020-11-11 07:38:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"958901ae8baaae7173cda18b1651d5e7c5ef159e","author":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"date":"2020-11-05 02:29:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dbd4b59fc679328a19fc4f3f5b0de849d28286e","author":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"date":"2020-11-05 02:37:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7e006c6386c6af122196500f6edbb29da0598aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 03:20:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba1770ea46e54a9b9f5af04a51c173a2 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a060864b5bcc44d0855ea0a6ce95549b : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cda075dd5fd74ed58db5519f3064d352 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c9c27166c84415db02a5e8dc119905d : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd70fc587c43404a8a1557c780a1988c : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/106e0f86233346f894a23d16ee262abd : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a321a5d6e6954ac1a25a947f2cc89d20 : SUCCESS in 3m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6328917098dc486699913cf2f69c60a1 : SUCCESS in 4m 49s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/6fe64322e12048ac9e28fcfe5ecfb63d : SUCCESS in 36m 17s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/d6091274a75247a99bf9c16fe7167999 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":2},{"id":"69da895ead76a8058408fe758d283e13c094ba0c","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2020-11-10 07:17:09.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM.\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"7933be14a1c80050de86027e2a363e6e2b95ebd9","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2020-11-11 06:34:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"23ff81bfd1191eeb1fd7a88027d044e94acd2077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 06:34:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f04c3775a3adea0109c9e205a3fd53206c6bf29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 07:38:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/006b789f5d78416b9951a09e42183584 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694946e6c6714d229d3631d1975d760c : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23dd4371e2574816828a1e96762ada66 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e43d24789e3a4912b3c47ad666c07e6b : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a8e949a2754927b03f50f3fa9c623d : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ebefb157cb4714822c0577a1099042 : SUCCESS in 3m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3db926a456434676a31b4831f4f286d1 : SUCCESS in 4m 33s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/aeca07549a304cd38ea24772a87a1438 : SUCCESS in 36m 12s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/001eb93d85ff4976af4b1fee99d35606 : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c9196d882896c3c99e63e7e11e676b2ee62c1da6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 07:38:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"99ea8d3452d699f89d3c192d1f127ecd6899a46c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 07:39:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2764313c994d46ff969a4a7e9455a06e : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b39f68827f4ac882237e1bff2e8f32 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e035439df743d8885d3f8f95299b8fa261746257","revisions":{"69b377fa99413018d44b922e5fe6de7c12edb09b":{"kind":"REWORK","_number":1,"created":"2020-11-05 02:29:04.000000000","uploader":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"ref":"refs/changes/88/761488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/88/761488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/761488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/761488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/761488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/88/761488/1"}}},"commit":{"parents":[{"commit":"cf196fc96906482aef18c54ca03378d35bc85e95","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cf196fc96906482aef18c54ca03378d35bc85e95"}]}],"author":{"name":"kuangcx","email":"kuangcx@inspur.com","date":"2020-11-05 02:23:59.000000000","tz":480},"committer":{"name":"kuangcx","email":"kuangcx@inspur.com","date":"2020-11-05 02:23:59.000000000","tz":480},"subject":"The docs  requirements migrated to doc/requirements.txt we need not install things from requirements.txt.请为您的变更输入提交说明。以 \u0027#\u0027 开始的行将被忽略，而一个空的提交","message":"The docs  requirements migrated to doc/requirements.txt\nwe need not install things from requirements.txt.请为您的变更输入提交说明。以 \u0027#\u0027 开始的行将被忽略，而一个空的提交\n\nChange-Id: Ia178c8fb54ec88e40af9fd63d00ea3d12b44c7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/69b377fa99413018d44b922e5fe6de7c12edb09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/69b377fa99413018d44b922e5fe6de7c12edb09b"}]},"branch":"refs/heads/master"},"e035439df743d8885d3f8f95299b8fa261746257":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-05 02:37:59.000000000","uploader":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"ref":"refs/changes/88/761488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/88/761488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/761488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/761488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/761488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/88/761488/2"}}},"commit":{"parents":[{"commit":"cf196fc96906482aef18c54ca03378d35bc85e95","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/cf196fc96906482aef18c54ca03378d35bc85e95"}]}],"author":{"name":"kuangcx","email":"kuangcx@inspur.com","date":"2020-11-05 02:23:59.000000000","tz":480},"committer":{"name":"kuangcx","email":"kuangcx@inspur.com","date":"2020-11-05 02:37:23.000000000","tz":480},"subject":"The docs  requirements migrated to doc/requirements.txt we need not install things from requirements.txt.","message":"The docs  requirements migrated to doc/requirements.txt\nwe need not install things from requirements.txt.\n\nChange-Id: Ia178c8fb54ec88e40af9fd63d00ea3d12b44c7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e035439df743d8885d3f8f95299b8fa261746257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e035439df743d8885d3f8f95299b8fa261746257"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
