)]}'
{"id":"openstack%2Ffreezer~764634","triplet_id":"openstack%2Ffreezer~master~I98ee2b16e65ed7a9a13e9f5193c90d588bf98375","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"I98ee2b16e65ed7a9a13e9f5193c90d588bf98375","subject":"Update TOX_CONSTRAINTS_FILE","status":"ABANDONED","created":"2020-11-30 11:05:30.000000000","updated":"2021-02-19 05:40:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"302654a4ba25785d278a69366b519d685a15e617","_number":764634,"virtual_id_number":764634,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-28 03:42:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 11:56:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 03:18:35.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"},{"updated":"2021-01-25 07:45:03.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"a373cce54e0ba8725435da3116f35e096a5759c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-30 11:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2726edd31f17807b612d0291bbed0f7fcf3ef35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 11:56:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e825c50b80149d9ae5f1866ace5756f : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a066f71797a41358873d59f857a2144 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2afff4052d74cc48eb073b016fb7828 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af3643f8bdc1440183e49632535c7c33 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/809673e7d3d943c8a19307ba88daca3b : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/257288af9cb04a44abd68026c315881e : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dd795706ee941748cd4fd3a4968fca1 : SUCCESS in 4m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/347f51c91dad4d1f8b76c148df16c29f : SUCCESS in 5m 39s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/7639a36fabfc4ccda69e8c94299f0dc4 : SUCCESS in 38m 47s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a93a3ca908b34d9e8b5d1a8f15bd1cd5 : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6c15ccd85c526bc3c19562c4ca4ed6de434828a0","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-01-25 03:18:35.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM.\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"f3ff46b7c8750aed4b900f17f1acf4248156ccd0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-01-25 07:45:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"09557c59f1d135717a3c76e1020116e05d441129","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-01-27 02:39:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81d2ef106bd77e9882cd46c830faf2a03c265428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:40:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b636cdd841c52e8d90f580ba8837bc39d4f0b48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 03:11:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3f02b1e55be4bd082bfada4415e8af6 : FAILURE in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1ddf4cc511455caddf4673d54dd9cc : FAILURE in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42bdc1bd8327449bbf8d42cb539d3ebe : FAILURE in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e62226205e04014a85960f61d589474 : FAILURE in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48651f176cfb4534870f932c02968395 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/336f0d79e088478d9499f353db636139 : SUCCESS in 4m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f5bd333e68146d99a5c6b36f946d1de : FAILURE in 3m 32s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ff34dcb3c49d4fdaaa15e70700d3a16c : SUCCESS in 29m 55s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/98c246506cc34ad1be624b15de5a1a09 : FAILURE in 28m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8cb6b7d4c1f3835338cf51843715d900af711fa5","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-01-27 05:49:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"34f7854829eb30050011db7bcbb86a8882ceff82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 06:30:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f6cbb6d99074973b40e7f7d3d8ec2fc : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e95d0f16d9a499f9de368fb37dbd032 : FAILURE in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a185976c537d488a8c71540c0ae7a842 : FAILURE in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dd02d952f92463b903d4fa08062d9b2 : FAILURE in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a0c2736c4a24f9cbaa2971adc4f21cf : FAILURE in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641ba788f1aa44d49524efeb4bb88a77 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b309219d71234297b231dc4ff7daf8c9 : SUCCESS in 3m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c5f78ed986b8486a9761b965f6d08750 : FAILURE in 3m 44s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/8f848f1420234857b145e565eed1fa4c : SUCCESS in 38m 46s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/0c0756d1e42545af87abe83a8ae28b9b : FAILURE in 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0360e20107da9e2b2b0e859fcac93a95a232388d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-01-28 02:52:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17cb1ae9114260c38e20c5a9e28f252484449289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 03:42:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8feefea051d74c4898b6d6372a27232d : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1af106772b6b434782f001906de5a10f : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a922f0dc394f27b25ada054b302762 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5153573814840dfa4821260a0afb8b1 : FAILURE in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6f44c307c584b5aa7e2168439b1091f : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22566656403b4ca1bed1ed35a66a5371 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8e48a9480a4f9bb5b3055f8d5e94f8 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0180b17b6c14e1f940c8701c26d279f : SUCCESS in 3m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f950637435ff4c3cb12f9871bd949234 : FAILURE in 5m 32s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/8758f04ceb5b4c17a1975c651a84891a : SUCCESS in 33m 56s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b341a4a3f3b14015a2d34856ca3ef7c6 : FAILURE in 35m 09s","accounts_in_message":[],"_revision_number":2},{"id":"302654a4ba25785d278a69366b519d685a15e617","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-02-19 05:40:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5843be8b896544ce0a68b1b5ae4da87150a4262e","revisions":{"28ed8047b984f3c61b76ee72103bb0ffb6b32291":{"kind":"REWORK","_number":1,"created":"2020-11-30 11:05:30.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/34/764634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/34/764634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/764634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/764634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/764634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/34/764634/1"}}},"commit":{"parents":[{"commit":"d4603d9e771bb07193344b065bc5f104eb17c018","subject":"Merge \"Dep\u0027s should be restricted by upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d4603d9e771bb07193344b065bc5f104eb17c018"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-30 11:04:31.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-30 11:04:31.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I98ee2b16e65ed7a9a13e9f5193c90d588bf98375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/28ed8047b984f3c61b76ee72103bb0ffb6b32291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/28ed8047b984f3c61b76ee72103bb0ffb6b32291"}]},"branch":"refs/heads/master"},"5843be8b896544ce0a68b1b5ae4da87150a4262e":{"kind":"REWORK","_number":2,"created":"2021-01-28 02:52:27.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/34/764634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/34/764634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/764634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/764634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/34/764634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/34/764634/2"}}},"commit":{"parents":[{"commit":"d4603d9e771bb07193344b065bc5f104eb17c018","subject":"Merge \"Dep\u0027s should be restricted by upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d4603d9e771bb07193344b065bc5f104eb17c018"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-11-30 11:04:31.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-01-28 02:52:14.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I98ee2b16e65ed7a9a13e9f5193c90d588bf98375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5843be8b896544ce0a68b1b5ae4da87150a4262e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/5843be8b896544ce0a68b1b5ae4da87150a4262e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
