)]}'
{"id":"openstack%2Ffreezer~774183","triplet_id":"openstack%2Ffreezer~master~I1abb4eca21a19ee810cf3017116eaf446b94af8b","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"I1abb4eca21a19ee810cf3017116eaf446b94af8b","subject":"add test_stopstate_stop for unit test","status":"MERGED","created":"2021-02-05 01:48:14.000000000","updated":"2021-02-05 02:56:37.000000000","submitted":"2021-02-05 02:55:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774183","meta_rev_id":"48183ba7f71ac5c8ea3bd2dabd763606de085bb7","_number":774183,"virtual_id_number":774183,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 02:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-05 02:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-05 02:21:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 02:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba7f9ac46d0ed86921fd019809ba05866137193b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-05 01:48:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ed4b38026063be8693ba65a3ef75bb27178813c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-05 01:51:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5aee31aea7bed4abd0eb38e2980d9fb1e72b1c8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:19:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/552f634701eb41cdbde9da6c20279103 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce04e6e9d6e44008b32cbe268bc57f8e : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bdff766224c456cb6d18a09f2d721a6 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ff0f1f0135c4c4bbe4eef24a4bc98b3 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddebed477ee74b64aef0c1a2311f1f9b : SUCCESS in 4m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9761c96392754a56aa0ec380f98b353a : SUCCESS in 4m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/36286e2d823148ccb164fc577e59bfb8 : SUCCESS in 27m 08s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/337c1f297b1842c5b3228d5414d6570d : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":2},{"id":"daec3008e73515ad930d6a9f3b89e029d35fb14c","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-05 02:21:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3bd5bb004cb4221e5844ee116fa77596700e1a5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:21:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"05ae11c8246feda5ccc755a3fcac5cb2d3ae8643","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:55:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d5d00eae4b540bfb9eb75c6ba38ec0a : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a498da3c6b44a16a652bd840e44e617 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b009e79bdab492d9dbce6af2485ff4c : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84636030ab364a8fa3f5b3604e9a58cd : SUCCESS in 4m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37e6fda00cb74b25bf103d1b2c51a15f : SUCCESS in 5m 29s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/56f0f44624494a648869946ffae52cb8 : SUCCESS in 29m 15s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/8be717c7c7fa4787a9e7d9e1a5df222b : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":2},{"id":"1a56584d22a8c3d0ea9bd832a2becd3294fd7e73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:55:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"48183ba7f71ac5c8ea3bd2dabd763606de085bb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:56:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28737f8bb3e64c4992ad8865bebab1a7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04023a00920e000c7d598677214454494b037575","revisions":{"e6977054469261df149a1736ba9782aa5b2b4678":{"kind":"REWORK","_number":1,"created":"2021-02-05 01:48:14.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/83/774183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/83/774183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/83/774183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/83/774183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/83/774183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/83/774183/1"}}},"commit":{"parents":[{"commit":"6e0bed59269f1ee6a978b5a6d8299024d8c48f10","subject":"add test_terminate_subprocess1 for unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6e0bed59269f1ee6a978b5a6d8299024d8c48f10"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 01:47:21.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 01:47:21.000000000","tz":-480},"subject":"add test_terminate_subprocess1 for unit test","message":"add test_terminate_subprocess1 for unit test\n\nChange-Id: I1abb4eca21a19ee810cf3017116eaf446b94af8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e6977054469261df149a1736ba9782aa5b2b4678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e6977054469261df149a1736ba9782aa5b2b4678"}]},"branch":"refs/heads/master"},"04023a00920e000c7d598677214454494b037575":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-05 01:51:28.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/83/774183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/83/774183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/83/774183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/83/774183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/83/774183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/83/774183/2"}}},"commit":{"parents":[{"commit":"6e0bed59269f1ee6a978b5a6d8299024d8c48f10","subject":"add test_terminate_subprocess1 for unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6e0bed59269f1ee6a978b5a6d8299024d8c48f10"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 01:47:21.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 01:50:38.000000000","tz":-480},"subject":"add test_stopstate_stop for unit test","message":"add test_stopstate_stop for unit test\n\nChange-Id: I1abb4eca21a19ee810cf3017116eaf446b94af8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04023a00920e000c7d598677214454494b037575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04023a00920e000c7d598677214454494b037575"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
