)]}'
{"id":"openstack%2Ffreezer~774188","triplet_id":"openstack%2Ffreezer~master~I850f92470380de21b91406eda28ca01339b61d43","project":"openstack/freezer","branch":"master","topic":"improveUTC","hashtags":[],"change_id":"I850f92470380de21b91406eda28ca01339b61d43","subject":"add test_stopstate_abort for unit test","status":"MERGED","created":"2021-02-05 03:07:13.000000000","updated":"2021-02-05 04:23:08.000000000","submitted":"2021-02-05 04:21:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774188-improveUTC","meta_rev_id":"b687fcc994549aa33550b860a5e9e763ec3df6fc","_number":774188,"virtual_id_number":774188,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 04:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-05 03:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-05 03:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 03:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec9357bf05c93dca3e142363dbb2c14fdd56d30d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-05 03:07:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3911df29c3fa1697c262083953ef37893254bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 03:50:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/411c2bd9e8a0404b95e3714b302a92ac : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/229a65f3e0e346ba971ac407daf3709f : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efc82b62ac3948c8ac6ef779cf998d87 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f8b1e608a184a52a41b0bc8a7932d86 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df07dd1e5e44fdebbbe8386f86f555a : SUCCESS in 4m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32c3691ebe2e46ebb8112a1084b305ef : SUCCESS in 4m 14s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/26a9e460249642f7ab19f2661204174c : SUCCESS in 28m 06s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ad794323a1164d8db72848ed33f1b03c : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"04211de3352824dcdb80c4ec3f570941e2d54f39","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-05 03:53:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94c921b851a17b499f9aa0e38ef0ac90807bce28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 03:54:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fadbba1eaa659e636daeef1f1038cb936e954d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 04:21:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d311b9a397a74b07978e71cb685aa7f5 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4385d620cc9e4292a3636fdffb435acb : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af0cddce89f46449265faeed0bcf926 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0aa34da048b4ff7828b1147907a8ff4 : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dafbb85f21d44df78bd2db10ada0a878 : SUCCESS in 3m 59s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/108759e267c941f89ce6e625ec6f252a : SUCCESS in 25m 35s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/098a1ed7e568426b8d754b6bed2703a6 : SUCCESS in 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f205d2b36b3f81bd8dc0dc22db9db626f5750723","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 04:21:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b687fcc994549aa33550b860a5e9e763ec3df6fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 04:23:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538045d556f54133825c249a4137dd3f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50885c1041c5ad4fa304d84dc009ba91c8801545","revisions":{"50885c1041c5ad4fa304d84dc009ba91c8801545":{"kind":"REWORK","_number":1,"created":"2021-02-05 03:07:13.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/88/774188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/88/774188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/774188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/774188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/88/774188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/88/774188/1"}}},"commit":{"parents":[{"commit":"04023a00920e000c7d598677214454494b037575","subject":"add test_stopstate_stop for unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/04023a00920e000c7d598677214454494b037575"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 03:06:38.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 03:06:38.000000000","tz":-480},"subject":"add test_stopstate_abort for unit test","message":"add test_stopstate_abort for unit test\n\nChange-Id: I850f92470380de21b91406eda28ca01339b61d43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/50885c1041c5ad4fa304d84dc009ba91c8801545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/50885c1041c5ad4fa304d84dc009ba91c8801545"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
