)]}'
{"id":"openstack%2Ffreezer~774206","triplet_id":"openstack%2Ffreezer~master~Iccd20901384ccee82dbb1e57dc6dddf211cbbcd0","project":"openstack/freezer","branch":"master","topic":"improveUTC2","hashtags":[],"change_id":"Iccd20901384ccee82dbb1e57dc6dddf211cbbcd0","subject":"add test_scheduledstate_start for unit test","status":"MERGED","created":"2021-02-05 08:44:07.000000000","updated":"2021-02-05 10:09:58.000000000","submitted":"2021-02-05 10:08:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774206-improveUTC2","meta_rev_id":"661a8ac25fff4387f96b198d63f39fb415c0725d","_number":774206,"virtual_id_number":774206,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 10:08:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-05 09:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-05 09:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 09:21:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5ced83a7d4bd46814aa166c4e2b0990d911f012","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-05 08:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa686ec894a91da6d6cbfb7aa2afd13261b45ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 09:21:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d0604d9e8945c68ff0ed0f8a542080 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f743b9e271ba49d9a8b766e59c8aff13 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/618e93b57e3f4e1691c950b28b2ee030 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/630f23b0c67445ffb77dc417bea60bfe : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbca4aa382cc44699d9a18972edb5b02 : SUCCESS in 4m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5805b9a0c834d7489d5803b3d97f0a7 : SUCCESS in 4m 36s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a31fc857adf74becbc3bb677f7a410e4 : SUCCESS in 32m 07s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/fdf2f39dc1304b9f9324cde5830237fa : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c4d0a855ddc17afa0b95e909116851972588e5cf","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-05 09:32:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62bc670d1299354ae470e40ecc2ca276a724d81a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 09:32:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe7772ada199cb7c2505c3c2cdc7fa0b2987916e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:08:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7bda565b38546e4a2bffed2188623ee : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3367982cad54ef2948441caafa1c530 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/662fe978ebe44f0bac3125c22b66ad07 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fab418a1cb24db48d6db0111c685d9e : SUCCESS in 4m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c644b0643c7441e28c711f2079d2bad9 : SUCCESS in 4m 33s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/6572dee4319e4de1b2168d3749e0c0b1 : SUCCESS in 30m 15s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/d192a6ef7f864de8b403a153c0bc8812 : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":1},{"id":"34c1fcff91a74c84f6e764a259f84171c290d180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:08:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"661a8ac25fff4387f96b198d63f39fb415c0725d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe285844d40480e92609e8f59945c72 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97984c71d69ba4e05632c920fa62c1e392650a20","revisions":{"97984c71d69ba4e05632c920fa62c1e392650a20":{"kind":"REWORK","_number":1,"created":"2021-02-05 08:44:07.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/06/774206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/06/774206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/06/774206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/06/774206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/06/774206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/06/774206/1"}}},"commit":{"parents":[{"commit":"0ea6b00494cdb19826c0d4124970340dc67ce7a4","subject":"add test_scheduledstate_stop for unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0ea6b00494cdb19826c0d4124970340dc67ce7a4"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 08:43:18.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-05 08:43:18.000000000","tz":-480},"subject":"add test_scheduledstate_start for unit test","message":"add test_scheduledstate_start for unit test\n\nChange-Id: Iccd20901384ccee82dbb1e57dc6dddf211cbbcd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/97984c71d69ba4e05632c920fa62c1e392650a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/97984c71d69ba4e05632c920fa62c1e392650a20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
