)]}'
{"id":"openstack%2Ffreezer~774361","triplet_id":"openstack%2Ffreezer~master~I33388d127121cc8e017f9db7209ad03773098929","project":"openstack/freezer","branch":"master","topic":"improveUTC3","hashtags":[],"change_id":"I33388d127121cc8e017f9db7209ad03773098929","subject":"add test_runningstate_stop for unit test","status":"MERGED","created":"2021-02-07 00:11:10.000000000","updated":"2021-02-07 01:44:37.000000000","submitted":"2021-02-07 01:43:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774361-improveUTC3","meta_rev_id":"dea7c268985203bd1169c47e0e0e2a67ae7dd628","_number":774361,"virtual_id_number":774361,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-07 01:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-07 01:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-07 01:07:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-07 01:03:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44349ebf0852d7e675a4083221db6d183edd396d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 00:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23190c3747ba28ae9ba1346e2973e4647ed2067a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 01:03:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bc5120a9b34414952b6f12ec754f9e : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1154ef665c074581ab65b8bbf0784d98 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/044f16c30a4a43728862f38cbe6287ac : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dac3ac51d1d4e3184d22e927b8f3c64 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04aab22baf114d9084dd4de473bb326d : SUCCESS in 4m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b612e5cc3c14685a11fbe8188724ff2 : SUCCESS in 4m 18s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/b497b05041624aeca3332ba193fa8aca : SUCCESS in 39m 25s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/47252866f9494e5abce34fe76fa13bbd : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"08f60db0d9c34a3900555ba2e45d4bfac243cc69","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 01:07:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49f6088e2d7166a509b0588cdb2faefcff93f177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 01:07:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b285cf66b26af1eac7564f4cfa3426e87ad5e92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 01:43:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1b761e8e3748a5ba23b6eabcf14611 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58fe000c2b934ad0a8fcc353fb65952f : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d98de94c88ce49d3b94c852267e44832 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2cdc56f9ee41f0ad905bd7b9831aeb : SUCCESS in 4m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9cec936c0e6e40e0841039ce719db4d0 : SUCCESS in 4m 23s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/4617545b16f54b37af15e0cc789abdf0 : SUCCESS in 28m 15s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e2f302a52e974fbb857c4fa251748a57 : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":1},{"id":"603242a2155db779f99d426c3f251fc6c5010f93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 01:43:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dea7c268985203bd1169c47e0e0e2a67ae7dd628","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 01:44:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4870807810446aa8912428bd170c3ed : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b26e2738bf7abac57497af896dcac896caf3b16","revisions":{"8b26e2738bf7abac57497af896dcac896caf3b16":{"kind":"REWORK","_number":1,"created":"2021-02-07 00:11:10.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/61/774361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/61/774361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/774361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/774361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/61/774361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/61/774361/1"}}},"commit":{"parents":[{"commit":"97984c71d69ba4e05632c920fa62c1e392650a20","subject":"add test_scheduledstate_start for unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/97984c71d69ba4e05632c920fa62c1e392650a20"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 00:10:40.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 00:10:40.000000000","tz":-480},"subject":"add test_runningstate_stop for unit test","message":"add test_runningstate_stop for unit test\n\nChange-Id: I33388d127121cc8e017f9db7209ad03773098929\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8b26e2738bf7abac57497af896dcac896caf3b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8b26e2738bf7abac57497af896dcac896caf3b16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
