)]}'
{"id":"openstack%2Ffreezer~774363","triplet_id":"openstack%2Ffreezer~master~I3db2a946d0950c9b01ee8912c43f7704ffd97c5d","project":"openstack/freezer","branch":"master","topic":"improveUTC4","hashtags":[],"change_id":"I3db2a946d0950c9b01ee8912c43f7704ffd97c5d","subject":"add test_runningstate_remove for unit test","status":"MERGED","created":"2021-02-07 01:48:42.000000000","updated":"2021-02-07 03:08:23.000000000","submitted":"2021-02-07 03:07:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774363-improveUTC4","meta_rev_id":"2c83f5e450342030cf8e1abfea0e11d6c28b8a93","_number":774363,"virtual_id_number":774363,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-07 03:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-07 02:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-07 02:27:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-07 02:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72462819e0f0d5e69bca4eced9a6aaa130fbd1d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 01:48:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc2fb1e000b9a022a299cd68833a482619f8a5d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 02:26:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f7e041e9884f8d829f2343bdc94363 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd9309f8d894314b740b4f03f5dd431 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee1b4eb898324d7b9bac50777eb1fc0e : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66dca79440cd48e686e0145bc1304a18 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a163b6e5114f1f9cd088a51a260cf7 : SUCCESS in 4m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/147701f665154e279fd5a769dcf54764 : SUCCESS in 4m 35s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/bb6efa982c0643d29198f4e836f866ff : SUCCESS in 37m 05s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a721ea3710b44a2f9d839e9e09315bb0 : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":1},{"id":"43caa273eba619702e8c02e9cb18b6e552324bae","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 02:27:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0bd940d6da21721b8a2390ff7dc1d5f5b0f7015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 02:28:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69633b37c1087093b418fb9d3ea7c470626f9f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 03:07:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc7fea69b4a7402681b79c13d0eb6ecb : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/680035be08784df8a4d3bfe73b6da682 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e0e3292af54556a3e1febd2d4baa83 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442a97203a28465199d0b8e7ca88b060 : SUCCESS in 5m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7da77654a2048928f28160a548687f9 : SUCCESS in 5m 42s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/7f4a04645a3a440097150038fb8385ae : SUCCESS in 35m 12s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a1a47745746a482da1833817a48571f8 : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ea30aaf64f51441964c224687a31d08e6fd9f3b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 03:07:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c83f5e450342030cf8e1abfea0e11d6c28b8a93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 03:08:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8bbe708b18d48029d7b063bb35cd73b : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97236a160994f139ca1302bb7cd5d68bacc7c15f","revisions":{"97236a160994f139ca1302bb7cd5d68bacc7c15f":{"kind":"REWORK","_number":1,"created":"2021-02-07 01:48:42.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/63/774363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/63/774363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/63/774363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/63/774363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/63/774363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/63/774363/1"}}},"commit":{"parents":[{"commit":"8b26e2738bf7abac57497af896dcac896caf3b16","subject":"add test_runningstate_stop for unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8b26e2738bf7abac57497af896dcac896caf3b16"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 01:48:01.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 01:48:01.000000000","tz":-480},"subject":"add test_runningstate_remove for unit test","message":"add test_runningstate_remove for unit test\n\nChange-Id: I3db2a946d0950c9b01ee8912c43f7704ffd97c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/97236a160994f139ca1302bb7cd5d68bacc7c15f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/97236a160994f139ca1302bb7cd5d68bacc7c15f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
