)]}'
{"id":"openstack%2Ffreezer~774364","triplet_id":"openstack%2Ffreezer~master~I8955f673b968c34b9b27f6e305f17ab3b6fcfac2","project":"openstack/freezer","branch":"master","topic":"improveUTC5","hashtags":[],"change_id":"I8955f673b968c34b9b27f6e305f17ab3b6fcfac2","subject":"add test_job_create unit testcases","status":"MERGED","created":"2021-02-07 03:14:21.000000000","updated":"2021-02-07 05:59:35.000000000","submitted":"2021-02-07 05:58:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774364-improveUTC5","meta_rev_id":"ec5f410bd2b2d1bfcadef5be014658e6c8134604","_number":774364,"virtual_id_number":774364,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-07 05:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-07 05:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-07 05:21:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-07 03:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57c7af8f8c1046f6387e13e4c2c2e1238c3d24e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 03:14:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a20eedf2d37048cb7f0e00c9e25420b0818ddb1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 03:48:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e6c253827e44f9a969286d93efca4b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79d78b94b45d42368eabe84d26a9dcf0 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed44d97abd07481cbeac0c8c7938bde2 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57bc02040c094fef841f7b920df55420 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a7237d373f40fbbcdf93dccf3df8ff : SUCCESS in 4m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/919d1c39b129472db7516cf02b444012 : SUCCESS in 4m 33s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/dec622d4fcb94b96b29ae615f11eecf8 : SUCCESS in 31m 25s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1a385665bce842a79990dee7b3f5a894 : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6c6f0d8801ffc8cc706ddec1ce3ee8cad136c03d","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 05:21:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a81816d485ce100fb80320e2913cefebd67887be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 05:21:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ce52f0a7e10b8d06ff957cfebdc46fa7f999d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 05:58:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5691eef7d0df4b75ad1f1dbc99779f78 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/093fa4027f5d434c8b51c4142485f03e : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd94731970dd45dab42b39641f0fcbbb : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23ba13eda9940e9b4ce632404328924 : SUCCESS in 5m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f7427c322524504a75f4712c1b0550c : SUCCESS in 4m 17s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/0ef0f697d431458e8be4b1d7126c4196 : SUCCESS in 36m 23s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/3b92e7aff4e542eab7c25d3015a49543 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4675cdad5cd01fdcf65d5327b331354aef4ff985","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 05:58:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec5f410bd2b2d1bfcadef5be014658e6c8134604","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 05:59:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72f9c64fbbe4dbeaa0f4b108baa47d3 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf49fb0dfc8bf2ab35921db46bf9e4878c5bad78","revisions":{"bf49fb0dfc8bf2ab35921db46bf9e4878c5bad78":{"kind":"REWORK","_number":1,"created":"2021-02-07 03:14:21.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/64/774364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/64/774364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/64/774364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/64/774364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/64/774364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/64/774364/1"}}},"commit":{"parents":[{"commit":"97236a160994f139ca1302bb7cd5d68bacc7c15f","subject":"add test_runningstate_remove for unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/97236a160994f139ca1302bb7cd5d68bacc7c15f"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 03:13:39.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 03:13:39.000000000","tz":-480},"subject":"add test_job_create unit testcases","message":"add test_job_create unit testcases\n\nChange-Id: I8955f673b968c34b9b27f6e305f17ab3b6fcfac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bf49fb0dfc8bf2ab35921db46bf9e4878c5bad78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/bf49fb0dfc8bf2ab35921db46bf9e4878c5bad78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
