)]}'
{"id":"openstack%2Ffreezer~774374","triplet_id":"openstack%2Ffreezer~master~Ifcef7876c3b1930ebd1237e9c1264db8278e332e","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Ifcef7876c3b1930ebd1237e9c1264db8278e332e","subject":"add test_job_process_event unit test cases","status":"MERGED","created":"2021-02-07 08:51:17.000000000","updated":"2021-02-07 09:55:54.000000000","submitted":"2021-02-07 09:54:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774374","meta_rev_id":"c245d8a275ff579d9014049b7777010a52e0b218","_number":774374,"virtual_id_number":774374,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-07 09:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-07 09:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-07 09:14:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-07 09:26:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07c00a1c499b6927bb7113f7729892cba040f350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 08:51:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db63b01a57cbe57403432ce8b9db1cef4f2945d6","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-07 09:14:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43480ee3826bac8509811ec283de1c948c5a099a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 09:26:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a846221cd5e495fbe4c1c472b3654e5 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e46d5c00fef4b03a9031cf4dc156c89 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6114ec29c3ae4d2c93a7c81d6a1921f5 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a106bdc8efd4dc4aa18cc67f6ac6297 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837089e7dbcf4bf4a9de62a110e0ffaf : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a02c61a99cb44f52907e50365087bb54 : SUCCESS in 4m 49s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d61cdf8f2ba24358b205f549cdfa7009 : SUCCESS in 28m 16s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/21eb4cdb7d324218973ec8e0c51fdc29 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f971dccc719c7a60c5afbd267d655eee8d6556c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 09:26:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c056b6397bf18c3fa2996f3dec3a50e115a9476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 09:54:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54cf428442143b78febeae6c6985647 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6034784d46c478685036672fdb4fdf1 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72d0fcb820cf4b1aaf30f91ad1bb2b5e : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dc76f920cd64a46bbbe115012b4994f : SUCCESS in 3m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf1e9cda2d994e108d4e5fdc5cf272de : SUCCESS in 4m 16s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d989c0d31b1842e4847f38a717da0ab6 : SUCCESS in 27m 00s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/88348bcc1f304b22b3522ca9f34ffcb8 : SUCCESS in 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c2ab15a13ad73068c7c46870b50a1e3164b8d878","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 09:54:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c245d8a275ff579d9014049b7777010a52e0b218","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 09:55:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb18ab85061d4b228e77c38ce7b105bd : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98b12b02e0ae653fda3034993afa6eaca42a3113","revisions":{"98b12b02e0ae653fda3034993afa6eaca42a3113":{"kind":"REWORK","_number":1,"created":"2021-02-07 08:51:17.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/74/774374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/74/774374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/74/774374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/74/774374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/74/774374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/74/774374/1"}}},"commit":{"parents":[{"commit":"ca74ddf145f92db30fc1ec9cc507d5b96b788cbb","subject":"add test_save_action_to_file unit test cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ca74ddf145f92db30fc1ec9cc507d5b96b788cbb"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 08:50:11.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-07 08:50:11.000000000","tz":-480},"subject":"add test_job_process_event unit test cases","message":"add test_job_process_event unit test cases\n\nChange-Id: Ifcef7876c3b1930ebd1237e9c1264db8278e332e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/98b12b02e0ae653fda3034993afa6eaca42a3113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/98b12b02e0ae653fda3034993afa6eaca42a3113"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
