)]}'
{"id":"openstack%2Ffreezer~774387","triplet_id":"openstack%2Ffreezer~master~Ia39021e543fea24f686caeacd8dcef964c80f431","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Ia39021e543fea24f686caeacd8dcef964c80f431","subject":"add test_job_start_session unit test cases","status":"MERGED","created":"2021-02-08 01:30:58.000000000","updated":"2021-02-08 02:29:18.000000000","submitted":"2021-02-08 02:28:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774387","meta_rev_id":"5c645fb46095491b3201a4a6796d867edc00e6e2","_number":774387,"virtual_id_number":774387,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 02:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":2,"date":"2021-02-08 01:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"all":[{"value":1,"date":"2021-02-08 01:32:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 01:58:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58bc8ed16daacab7b423ec1ed50e3e110f89cd72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-08 01:30:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a947c913b14300f89b8a72b0abd5fe9015daf37","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2021-02-08 01:32:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f58d2717fdac78413c458048ab631dea63b922f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 01:58:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d24aca8d287841cd8e0710b3e618dcf2 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b160578198fb4e50b70c6ab6cd8e3d91 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d78398b11544261bcd52d4ed3e86948 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99580239926545c4ace872ec550de03d : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f8d83f96a341d589e4d2da5f6d78a7 : SUCCESS in 3m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73b97f3f540643c982929d000864c74a : SUCCESS in 3m 57s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/f2248e136eee409f96112efbc7bb881f : SUCCESS in 23m 30s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/cef9b69a8e454621ae1819365bdc007f : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a56326beb3232b40a831c1690396dbc9c91492c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 01:58:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4161f70a5f3a3c6a787c4eb1ff0447691a4cd333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 02:28:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62c78d15ff34d1fa45ced0935cd1e32 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ace749c756b4a03912ecfe31a4ca300 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/184acab2d3aa4f888ec3c65ad4351d12 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cef6435c3864031b305b4b5467caff2 : SUCCESS in 4m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/066b5698b8dc4aa1bb52fc1fbb9049ae : SUCCESS in 4m 30s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/311725cee6a64f4e8f88bd4f81f963ce : SUCCESS in 27m 55s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/eeb4ccd190ff4866b34b9b6e8cac1adf : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ef74aa54116732808c58382fc6f0c287def7946e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 02:28:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5c645fb46095491b3201a4a6796d867edc00e6e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 02:29:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb9c9ec4dec4cbfaea148948e513276 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9773f99febb853ca9fc44756428ce1874ffdf0f5","revisions":{"9773f99febb853ca9fc44756428ce1874ffdf0f5":{"kind":"REWORK","_number":1,"created":"2021-02-08 01:30:58.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/87/774387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/87/774387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/774387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/774387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/87/774387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/87/774387/1"}}},"commit":{"parents":[{"commit":"71f04ce289871e69fcaa4c69ab9b7de17bd9faf9","subject":"add test_job_execute unit test cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/71f04ce289871e69fcaa4c69ab9b7de17bd9faf9"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-08 01:30:05.000000000","tz":-480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2021-02-08 01:30:05.000000000","tz":-480},"subject":"add test_job_start_session unit test cases","message":"add test_job_start_session unit test cases\n\nChange-Id: Ia39021e543fea24f686caeacd8dcef964c80f431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9773f99febb853ca9fc44756428ce1874ffdf0f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9773f99febb853ca9fc44756428ce1874ffdf0f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
