)]}'
{"id":"openstack%2Ffreezer~789040","triplet_id":"openstack%2Ffreezer~master~Ib6b5c10bb8e9179619275fd21c35d4e7936baaad","project":"openstack/freezer","branch":"master","hashtags":[],"change_id":"Ib6b5c10bb8e9179619275fd21c35d4e7936baaad","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-01 07:12:50.000000000","updated":"2021-06-09 11:22:51.000000000","submitted":"2021-06-09 11:21:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789040","meta_rev_id":"1499d56c05a8888458c37860cb5749911a76ec49","_number":789040,"virtual_id_number":789040,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 11:21:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 10:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-25 10:57:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-01 08:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 10:57:12.000000000","updated_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"reviewer":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"state":"REVIEWER"}],"messages":[{"id":"5fc8137b30e6777938435cb61f54efc13511f216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-01 07:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f3c8de6f2a2c8561ae3de7c3685618d624a0847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 08:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4472ddc01e1b44e4badf2a794c808ef4 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52d163c93db648ae93e417112c7aa1a1 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96381bcc72fa44999fc1f5be133901b0 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf70f9e4d864472789546e85a2511a60 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36f95d9341464ef5b21e660b72d89ebd : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5ef89f84d245a69f3806b1461745b9 : SUCCESS in 5m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/639af25537d6440cb3713accfe428250 : SUCCESS in 5m 20s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a5778fcc4ea348b79b3273d40dc0be0d : SUCCESS in 31m 02s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/8cf0d1b838334b3ca55ba94e3558d7a8 : SUCCESS in 26m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6566078a6eae92d8df1f759124bc0543c1fb822d","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-05-25 10:57:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM.\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"f58883e4abcaefcf4b61a4b097160f375c2007c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 10:57:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5133b17bff7f3ce63ebfcc89a7a9d6d1169065eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 11:04:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8890ba50d50649958316ccacdc89511a : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ab6430f5864e219eded28778c6358d : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2937d3e409d74eada278402fe398f16d : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64af70459b5943609d6d1343ea120953 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9535701a7064ea1a698f45e5e0b4e9b : SUCCESS in 6m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f865634367ea4a4c91fbe08e5ead6f94 : SUCCESS in 6m 04s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/2af964b418424f97adcb7847ab63b21b : FAILURE in 6m 35s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/98a6dc9ae6fa4531bc92401d348f1d14 : FAILURE in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"076fd6a877dfd464818049b3dc763fcbfc247990","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-26 00:46:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"246d75eacd616be5b38e2efdceb704eb25fc6426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 00:53:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66dd181e7d114478ad102cd2666f7acd : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c41ae93f93c44b0d86dd4622132f19c5 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33d071f37bc6455f9cbb70bde1c90668 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2feeabe60875437c87658fa15c25b97d : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c64b246a66b943568b66e3cba249c49c : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c7035450234b49bdcbf302260b92c7 : SUCCESS in 5m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b69ab207f2d2481ead12a9340e97e1e9 : SUCCESS in 5m 19s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/469815ed1e9d4b6387cb819acae91e3f : FAILURE in 5m 54s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e5bb90ffa36442e7a18faa547534ac80 : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"eed96e30ca5f4f9067447feeddcfa4e0584a3702","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-06-01 02:43:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef103d6bf6dad9780c5c308799440afe325d7c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 02:49:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3188355df1a54f3fb81ec9187e810cbb : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a365bc80804b59820734fcac85d8c3 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/647f74f502974f0383b8f9b3a1e01852 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/843cba67867940249c0e0e567113b8cd : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15003351f5304bb0a1c6d174594babc1 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e271f0bd34452a8481986bde502862 : SUCCESS in 5m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5784e607ebaf453081f6f58302fea980 : SUCCESS in 5m 52s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/61aef117b8e948a6a8a4996de377e0eb : FAILURE in 6m 15s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/daf34945b87d4fef9c35e80cdf6bed6b : FAILURE in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"05ca4f4bc80abea7b75deecac530fad6bd1777e6","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-06-06 23:40:55.000000000","message":"Patch Set 1:\n\nRecheck","accounts_in_message":[],"_revision_number":1},{"id":"4cbde9e25e7b4bf86eb0b8a8fdb94a9b36772945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 00:10:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7091b0055ac14c4094d49948fe9e43d5 : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a821a7fb2807409bb30240e478b82cdf : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f8099995e844cd95dbe1f68a9b351b : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a420068db84480092212905c0578ce5 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aa178bbb33e45f7b1e1a1ebf438145f : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8b9c58798f4227a8e5b6ef80ed2e55 : SUCCESS in 5m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbaaa3e3039a4c2c99eee70462036066 : SUCCESS in 5m 27s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/1a071c11a70a481695dcf25df9a2c852 : FAILURE in 28m 09s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a11ddc57e566400b99a5eb8c4cc0e150 : FAILURE in 28m 07s","accounts_in_message":[],"_revision_number":1},{"id":"19fe3eaa8522cd041da3508d1458dd48e253cf59","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-06-09 08:56:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d56488b3a83e2d2171a5c17a818f6f98a601e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:53:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d796ebbb487c4a88b652c40d9343bd1e : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93088452e69477f93394a8b7fe5661b : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0841d15e276c41f99863c935e649ae95 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c60cd2a681d4b61a499b4a13dfea860 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55540a1724944e0aafc9de595498d125 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e4397a01707425a8ded1e0ae9d503fb : SUCCESS in 5m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/479433b3687444cf83410baa3027720b : SUCCESS in 5m 13s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/20b4305a3d204c45936bc27a6c2b2d56 : FAILURE in 34m 22s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/aa7d36137f21411985cb731e32391bc0 : FAILURE in 51m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9ec9168f41998ffdd1e586208e1efbe526239f59","author":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"},"date":"2021-06-09 10:14:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ddc6b38a99dde8b8afb39f43d7fd8ab501bcde17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:50:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd44de06998c4145a9bdcc734f6725a6 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de10c908600146f29f9c63e448d3eccb : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7fdf05d9b964d5c998848ac929799fe : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/662194de7369406eb11906bef77f9d2b : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e1f687f57334f82bf7e7ce311d9229b : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf2afa70ce04ac29f4827acd339f994 : SUCCESS in 5m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b934ce0dc06e4923a0952f188619552a : SUCCESS in 5m 32s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e73d1ce0108047779dfc18046177d4b3 : SUCCESS in 30m 45s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/5b18f423c58c410a9dc1ae4a5000d4b0 : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3a575948f5039c60d94c0dcaada91043d6b32a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:50:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b52b00efdd316e7acbfdf7c29e332205dbf09a04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:21:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7baa75f5c2ae4f97beda2aa59903d075 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45e33266d1aa40c89baea0c3c5a535f9 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93a583d52bcf44e9a245a2b2f7fe994e : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5349f4166f594ddda94078b118938bc2 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e82dc1358af40cbbe445aca13988241 : SUCCESS in 5m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7befc23295ea465a80274f4a68415701 : SUCCESS in 5m 33s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/597d028c9ac745869b059306dff314a9 : SUCCESS in 30m 13s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b8fb17b122ee41ff8d470b62c7839517 : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":1},{"id":"393f55123e0a9544285e4ebf5262f32f7c0294c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:21:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1499d56c05a8888458c37860cb5749911a76ec49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:22:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7a7ac0d27341b1aad33db6c0aba6c4 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea30ed7deeaa31b4bbf51e0a2f7de12749b8f867","revisions":{"ea30ed7deeaa31b4bbf51e0a2f7de12749b8f867":{"kind":"REWORK","_number":1,"created":"2021-05-01 07:12:50.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/40/789040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/40/789040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/40/789040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/40/789040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/40/789040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/40/789040/1"}}},"commit":{"parents":[{"commit":"9773f99febb853ca9fc44756428ce1874ffdf0f5","subject":"add test_job_start_session unit test cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9773f99febb853ca9fc44756428ce1874ffdf0f5"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-01 07:12:43.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-05-01 07:12:43.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nResolves warnings like the following:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\nChange-Id: Ib6b5c10bb8e9179619275fd21c35d4e7936baaad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ea30ed7deeaa31b4bbf51e0a2f7de12749b8f867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ea30ed7deeaa31b4bbf51e0a2f7de12749b8f867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21069,"name":"Carl caihui","email":"caihui_nj@163.com","username":"caihui"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
