)]}'
{"id":"openstack%2Ffreezer~886035","triplet_id":"openstack%2Ffreezer~stable%2Fzed~I3c95bb48c7fdd0c752db121f1745f64a53b5a0e5","project":"openstack/freezer","branch":"stable/zed","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-15 06:54:46.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I3c95bb48c7fdd0c752db121f1745f64a53b5a0e5","subject":"Fix deprecated zuul queue syntax","status":"MERGED","created":"2023-06-13 22:57:44.000000000","updated":"2023-06-15 06:59:21.000000000","submitted":"2023-06-15 06:54:53.000000000","submitter":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886035-fix-zuul-errors","meta_rev_id":"32ed380674cc8464871cadb0907719f7554fc9c4","_number":886035,"virtual_id_number":886035,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-15 06:54:46.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-06-13 23:43:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-06-14 08:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-15 06:54:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 22:57:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-06-13 23:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-13 23:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-14 08:04:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"0384598907ef7a64e8c78689050d3a2d289f3319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-13 22:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"959e5a6abd555b911c06e453e07c0661665b76e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:03:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/464e828c79ea4a87bc88b0a7ff4d81ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a55c7cad58549d5a8258d14968dcfbe : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e39b368a1da4485fbbb93bfb18cb8eee : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db7a2adff8765037a9a37fe63b674dfa67e2dbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/148dfd4309a04e809c2c8257bffa0ded\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/07d71a82f26f449eac95eb50c396763b : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131497fb28aa43278f088cb97a5f4b27 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49b164ee3e594989a4c543e7ab010a56 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e68acd976e164f20a5063e4eb56bca2a : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a86a9f64569846ecbf521cd49b2b345c : RETRY_LIMIT in 1m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b77ee1239e4bc2a9230846d1e11b26 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fea05bb853b24c0e96fec2428a15244f : SUCCESS in 3m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/750cf360b790439cb6d4fd40d55ef83e : SUCCESS in 4m 58s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/7595c5597fe5448d811cfac988dfbd8b : SUCCESS in 38m 17s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/5d3d9fa5667448cea2bedbabd5527c2a : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":1},{"id":"44b72b8d9d2d00a5963b51488e01e02a3ffa8507","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-14 08:04:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3a2f9d741e286f2d5b9085d4f4267f287f39d7e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:54:46.000000000","message":"Patch Set 1: Verified+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93cc682255dc08f59306102db80c9e2cbef94818","tag":"autogenerated:gerrit:merged","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-15 06:54:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"32ed380674cc8464871cadb0907719f7554fc9c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 06:59:21.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1491f9aa8a942ea9b5a9602f4dcff2d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52a3198750941deb3287f36348d64ad : FAILURE in 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c7d8a3ab9cb51fd1d185ce084292e09837139c7","revisions":{"6c7d8a3ab9cb51fd1d185ce084292e09837139c7":{"kind":"REWORK","_number":1,"created":"2023-06-13 22:57:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/886035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/35/886035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/35/886035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/35/886035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/35/886035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/35/886035/1"}}},"commit":{"parents":[{"commit":"b6ee543de58728496d789b8a6afdbd45bf7354db","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b6ee543de58728496d789b8a6afdbd45bf7354db"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-23 12:53:10.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-06-13 22:57:40.000000000","tz":120},"subject":"Fix deprecated zuul queue syntax","message":"Fix deprecated zuul queue syntax\n\nManually edited for stable backports to only fix the zuul config error\nto avoid merge conflicts.\n\n1. Zuul deprecated (and removed) declaring shared queues at a pipeline\nlevel [1]. This patch fixes the syntax to make the gate work again for\nfreezer.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n\nChange-Id: I3c95bb48c7fdd0c752db121f1745f64a53b5a0e5\n(cherry picked from commit f6a599609239afc14569ceab9c8870386a3b1139)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6c7d8a3ab9cb51fd1d185ce084292e09837139c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/6c7d8a3ab9cb51fd1d185ce084292e09837139c7"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
