)]}'
{"id":"openstack%2Ffreezer~973219","triplet_id":"openstack%2Ffreezer~master~I38d4f89fd1700a40d83ceb28ebd644a6aa1095a8","project":"openstack/freezer","branch":"master","topic":"fix/incremental-flag","attention_set":{},"removed_from_attention_set":{"36848":{"account":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"last_update":"2026-01-16 19:49:02.000000000","reason":"\u003cGERRIT_ACCOUNT_36848\u003e replied on the change","reason_account":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}},"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-01-19 14:19:50.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-16 10:44:06.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I38d4f89fd1700a40d83ceb28ebd644a6aa1095a8","subject":"Remove inconsistent no_incremental option","status":"MERGED","created":"2026-01-13 11:56:51.000000000","updated":"2026-01-19 14:20:44.000000000","submitted":"2026-01-19 14:19:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973219-fix/incremental-flag","meta_rev_id":"d264eeb599502f64842a213d0eda31ea15e954be","_number":973219,"virtual_id_number":973219,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-19 14:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"all":[{"value":2,"date":"2026-01-16 19:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"value":2,"date":"2026-01-16 10:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-18 19:32:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"all":[{"value":1,"date":"2026-01-16 19:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 12:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-15 17:49:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-01-16 10:21:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-01-16 10:44:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"state":"REVIEWER"},{"updated":"2026-01-16 10:44:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},"state":"CC"},{"updated":"2026-01-18 19:32:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"1de5b5a791a15a43df85301010320ba450b9ea33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-13 11:56:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2a8a41f92dcd92cddfaa2100d14d9b2b0f3dfe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-13 11:58:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dd56c167fa55de1086a1ec06b1bb2b1688b08f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 12:37:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcca1996dae345c9a75c4284ccf9342e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb80b684c0af4eae86f15bd628334f37 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3336827af67c40b9b2431edcd40e3648 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/624defc238fb43a39c34380916443c3c : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bbea23926a849759587dc41aacc4ed0 : SUCCESS in 6m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3cc9ecf5d97f4553bf3ff4d1120c518c : SUCCESS in 3m 24s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/199454e093564d4ebc842907dddf9ae2 : FAILURE in 34m 18s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/dc877114f47144c6b4446160815f6752 : SUCCESS in 36m 00s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2120e9d687d344ce953e6bd684419c30 : SUCCESS in 33m 43s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c877c9153af544c2bdce96f7af22d469 : FAILURE in 22m 33s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/219d49f820914ad28ca41c65c922af25 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":2},{"id":"95144862a18f2d6762f65df4fbd7171f331286d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-13 13:20:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cd24889bca0c6d16ef8fd55b3d3568b0851e9b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 14:17:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04e2dc9d18d04e93b90edcc0598d8504\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375fc70b66284ac59849e072b90cdf1f : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b696e0fcde3c420bb687b6232aa03c3c : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a77afd83f46474482748e69bda2708a : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db2ac235c11549b084641706a432e360 : SUCCESS in 7m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a47834da117445d9a9928142c4fd265b : SUCCESS in 3m 58s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/00d9e53728df4006ba491b4201c132e2 : SUCCESS in 36m 03s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/3a5ddf8d37594d0fad2134907c6e04e4 : SUCCESS in 14m 25s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d18ac0a54066490aaa50ef3e6f6ef139 : SUCCESS in 29m 15s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/1fd9b35352f7404a89258b8b9fb0b661 : SUCCESS in 17m 37s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/41433e04f56d40e485288984e8f34893 : SUCCESS in 11m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3694354b103c9c2c2ab5b11db6d17f0697677830","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-15 17:49:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"586f1d49ab228ebc9b63c6fd3b25bbdebee7717c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 09:58:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"469e0b0e38f3f147f724522aa3a2f804fffcd68f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 10:21:09.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"08998b94c439f06394d08c60151fa5992873f4e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 10:39:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0a718c6693452ca4d2618e3d3073d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734bd40524c74972941afded1229c63d : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66850b9c54f6403f9d402fa6963136f1 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43464cbaeb164c32a721222fd09da3b0 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8cb9afc4ed2475798701be1482db0cb : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4e0dff59da5416c913009c8663c8cf3 : SUCCESS in 6m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5e75c4f59e6248dfb5babc332982aca3 : SUCCESS in 2m 30s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e1419f0eb57c4c34a1cb647bf0e01fdf : SUCCESS in 34m 45s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/b73c5c2cd4fd401b9a481ee866b98b77 : SUCCESS in 14m 28s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22908ecbab73409c811e76d3f75ba720 : SUCCESS in 35m 19s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/25de5092b4ca495c81f6c4135b9502fc : SUCCESS in 31m 20s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c08d281d4094accbe4da83f7409924e : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":4},{"id":"6b901b9fd35cfe4a93e04f64ef35df7cf1938320","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 10:43:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"acb42dbcbb7c947df5dc1f3fe6d8393d9db2fa19","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 10:44:06.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5fcd675d08b273019c206d2dcfeffa82452e66f","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 10:44:29.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9bc46f46b7ceee524392c1f75680a5a46342e047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 11:24:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/257a361d109141748b674adc89b6c9d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab423d321bb4886a734f65249db0fc1 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2471cf1bb08046a5b722705550f427b2 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30980e56135435191639071d31825bd : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/683825f861a44fd5bdc08ffb99e30053 : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee11c9a899c943ba89a8b5fe283a3ec9 : SUCCESS in 6m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c10a563fa2bf40b6a3ea0597414162a3 : SUCCESS in 4m 04s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/d0949883aea24b51a7656fa0f291cd4f : SUCCESS in 34m 02s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/419fa27ad76f49b3836fb97d985afeaf : SUCCESS in 32m 06s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15d38770d27b46a98099abc742966839 : SUCCESS in 14m 42s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/d6eeed4991c04e339fe3dc7433d3fb4c : SUCCESS in 36m 14s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f85b6cddfc44a468ef1014414f867b0 : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":5},{"id":"96a941d54f3abf826e7b9e1410499381c308988a","author":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"date":"2026-01-16 19:49:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4d81a13ab887031ac43e261e3b448a8e4901c61e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 19:49:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6ea26db8ab2d7430cfd62c0fb8268494a47dc2e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 20:22:31.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85797e317a64484d826e39db584b6026\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f0c89e7aa745f1aa9d849d3f298c86 : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da2604e0310c4af191e8dcd0bcc60a19 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55bf435ac7b147e0a729768262f30876 : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43e8f566addb45d3bf0def1218966720 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9a73b3a06bd4a179b2034d9ec3bd227 : RETRY_LIMIT in 1m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea92dd4f343848d3bb2c9341fa888fe3 : SUCCESS in 2m 24s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/63bf3ecc1f54445292c05a0726af9327 : SUCCESS in 16m 50s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/81e96890f6a94439b9ff9ef25c6109c7 : SUCCESS in 14m 44s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7883a2c540604d2db0faf2eeef29db05 : SUCCESS in 24m 20s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/abc0187b17a64acb8e33da916660cc92 : SUCCESS in 26m 10s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8087126f22041f5b473d77ffeca2ecf : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":5},{"id":"95e9949f60fdfe9fac376a6462774c90aa0cf586","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 19:32:25.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a2b3fd5e26777972a78b3afb471cf16df2caeb90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-18 20:07:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d4c821cd804426383c3db1fcd66d477\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa1898f97f140468d57f93fad888b34 : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4a3faf137d34e3ea46397b045bcb690 : SUCCESS in 1m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a230f930c74fcab6735b1c00e00bd0 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8785f6d119e449bcae000c1c6800f90e : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f27e483f6e9403283f5f8418563dff8 : RETRY_LIMIT in 1m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa650452d84243c09ae1bf9bf9ae9bce : SUCCESS in 5m 03s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/3252fcb62c5942a5990e8c4a46b7dfde : SUCCESS in 20m 44s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/c6603390fb4f400ea83718d8ed6bc8c3 : SUCCESS in 14m 07s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50d06dd45b334b6e8afd0f9378d67898 : SUCCESS in 33m 02s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/9012da7fe6964d8aab5d7b68bbcab5dc : SUCCESS in 16m 57s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/13f83bb7f469417ab3df3ba1e28302a3 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":5},{"id":"0937fb4749927af45cf48cc79499716b6fba3e29","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 08:02:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cad0ca544dd5cdf11092c02a9f8e3a3529174b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 08:44:33.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56f7d1d402744bcb8549d71f5514f34b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85109f4e7f324fe0b29097873cb196fe : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0d378a0a89a4154bfbc0d388628eec9 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853eb2944e064fcd86a13e33fc55b4d7 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3dcc0e4e19d841dca99ddc63ec3cbe47 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d64828403994e1f99fd9213fe6f1f10 : RETRY_LIMIT in 1m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f28a435526c64646a7a84bb4ebcf198a : SUCCESS in 5m 14s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/50899aa24d3c49c5b4cb1bd965413cfc : SUCCESS in 36m 46s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/c78ecde4eea9417ea724c74221d14e8f : SUCCESS in 36m 48s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/973220e64b9a4c818923be37615a67f4 : SUCCESS in 34m 47s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/df6498cf54a0475cad383447783c13a0 : SUCCESS in 36m 33s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/742ffeef954447d7b6ccc8441fe6f94c : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":5},{"id":"0cc66f8b18c136debae7f2b11076c04f8776c849","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 13:00:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"85bdd887e0d6a83ace4f8714b18a7f810fd0d9af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:42:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c29f4cff7c084c6d940ca2e8f2dbfda8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719e6c397073409d81e9c98e1b940ad5 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8b5cd158842499f92a3b4a0864996d4 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57c8c44becc40e98211b67d988042c3 : SUCCESS in 2m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e3bb5f0d9ca48b5b092b16d69e0e952 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b887d2d301ac40c0a33545a3b5e11395 : SUCCESS in 2m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b21d97cf193443d8cd40854018a9aaf : SUCCESS in 2m 40s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/33ddca87cd1b4d6f8d2035eb87f19723 : SUCCESS in 16m 05s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a8b181c9e8574b3b8f2b99c828eef020 : SUCCESS in 18m 09s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39063f1e66e24e398363a1f2be55f767 : SUCCESS in 36m 50s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/984f9e89001d4ee5af0304881db04617 : SUCCESS in 30m 07s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/99ce3ab4cb754e20a5aa5ba19574e048 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":5},{"id":"003bd0d0177d96e232252d41960166fa785de697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:43:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"88fbad09f36c4914f8d0eecfdbd9ef5eda289f6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 14:19:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edaf3630096948efa4ab08ff9fa3825d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c3e8c6435d4b3a8861024d33b826d7 : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62c8275a0dfb476c8f22929eebd46d52 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c20c92d65f466dba2fe6c63ae23385 : SUCCESS in 2m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11b57f6e0aad45d9bb9da21428330085 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d54c94f556246bda8f6301403fa41b1 : SUCCESS in 2m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd8e0660ad5b42a7b4ae43fff4f0b74f : SUCCESS in 2m 38s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/2b202d8577934e899378744aed50da39 : SUCCESS in 17m 14s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1e59b0029dce458c9f94a2a42f976bef : SUCCESS in 34m 17s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a42a62af193d469d8b5f3ea55aec5e32 : SUCCESS in 17m 19s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b98ddee6547244e6b918a9c47ef9775c : SUCCESS in 35m 35s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/446b8f47d52c4b45893f539ccded360e : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b0242a58954b7af6bbc8cd4a3e804735abae7989","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 14:19:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d264eeb599502f64842a213d0eda31ea15e954be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 14:20:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/358aa846031c405db4d074e5ee4fc475\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e5d82789f0403fabf14dcafc51eb72 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eadb6fa68d414951a8d870c4bca61480 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a55c162ea8c96903850f3eda95ce79ace240e7b9","revisions":{"85368783534c6c1db2d5463b1280644a3c85954c":{"kind":"REWORK","_number":1,"created":"2026-01-13 11:56:51.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/19/973219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/19/973219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/19/973219/1"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 10:56:04.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 11:56:49.000000000","tz":60},"subject":"Remove inconsistent no_incremental option","message":"Remove inconsistent no_incremental option\n\nThis removes the \"--no-incremental\" flag and \"no_incremental\" option and uses\nthe inverse of the existing \"--incremental\" flag or \"incremental\" option\nrespectively instead.\n\nIt fixes the issue that \"incremental\" defaulted to False while \"no_incremental\"\ndefaulted to None (effectively False) as well and both options not being\ncoupled so conflicting values were not caught. Parts of the code used one\noption while other parts used the other ones, leading to inconsistent\nbehavior.\n\nCloses-Bug: #2138294\n\nChange-Id: I38d4f89fd1700a40d83ceb28ebd644a6aa1095a8\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/85368783534c6c1db2d5463b1280644a3c85954c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/85368783534c6c1db2d5463b1280644a3c85954c"}]},"branch":"refs/heads/master"},"0be5b0f3ba99f33bcfb792b0700d7cf10ef34798":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-13 11:58:38.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/19/973219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/19/973219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/19/973219/2"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 10:56:04.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 11:57:41.000000000","tz":60},"subject":"Remove inconsistent no_incremental option","message":"Remove inconsistent no_incremental option\n\nThis removes the \"--no-incremental\" flag and \"no_incremental\" option\nand uses the inverse of the existing \"--incremental\" flag or\n\"incremental\" option respectively instead.\n\nIt fixes the issue that \"incremental\" defaulted to False while\n\"no_incremental\" defaulted to None (effectively False) as well and both\noptions not being coupled so conflicting values were not caught. Parts\nof the code used one option while other parts used the other ones,\nleading to inconsistent behavior.\n\nCloses-Bug: #2138294\n\nChange-Id: I38d4f89fd1700a40d83ceb28ebd644a6aa1095a8\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0be5b0f3ba99f33bcfb792b0700d7cf10ef34798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0be5b0f3ba99f33bcfb792b0700d7cf10ef34798"}]},"branch":"refs/heads/master"},"72e97874dd2eb663843dcd6180a70a7e4f259929":{"kind":"REWORK","_number":3,"created":"2026-01-13 13:20:07.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/19/973219/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/19/973219/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/19/973219/3"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 10:56:04.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 13:19:59.000000000","tz":60},"subject":"Remove inconsistent no_incremental option","message":"Remove inconsistent no_incremental option\n\nThis removes the \"--no-incremental\" flag and \"no_incremental\" option\nand uses the inverse of the existing \"--incremental\" flag or\n\"incremental\" option respectively instead.\n\nIt fixes the issue that \"incremental\" defaulted to False while\n\"no_incremental\" defaulted to None (effectively False) as well and both\noptions not being coupled so conflicting values were not caught. Parts\nof the code used one option while other parts used the other ones,\nleading to inconsistent behavior.\n\nCloses-Bug: #2138294\n\nChange-Id: I38d4f89fd1700a40d83ceb28ebd644a6aa1095a8\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/72e97874dd2eb663843dcd6180a70a7e4f259929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/72e97874dd2eb663843dcd6180a70a7e4f259929"}]},"branch":"refs/heads/master"},"de56476a0bb964d16dec206a784315fdf1c1eaee":{"kind":"REWORK","_number":4,"created":"2026-01-16 09:58:01.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/19/973219/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/19/973219/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/19/973219/4"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 10:56:04.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 09:57:43.000000000","tz":60},"subject":"Remove inconsistent no_incremental option","message":"Remove inconsistent no_incremental option\n\nThis removes the \"--no-incremental\" flag and \"no_incremental\" option\nand uses the inverse of the existing \"--incremental\" flag or\n\"incremental\" option respectively instead.\n\nIt fixes the issue that \"incremental\" defaulted to False while\n\"no_incremental\" defaulted to None (effectively False) as well and both\noptions not being coupled so conflicting values were not caught. Parts\nof the code used one option while other parts used the other ones,\nleading to inconsistent behavior.\n\nCloses-Bug: #2138294\n\nChange-Id: I38d4f89fd1700a40d83ceb28ebd644a6aa1095a8\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/de56476a0bb964d16dec206a784315fdf1c1eaee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/de56476a0bb964d16dec206a784315fdf1c1eaee"}]},"branch":"refs/heads/master"},"a55c162ea8c96903850f3eda95ce79ace240e7b9":{"kind":"REWORK","_number":5,"created":"2026-01-16 10:43:09.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/19/973219/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/19/973219/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/19/973219/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/19/973219/5"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-13 10:56:04.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 10:42:49.000000000","tz":60},"subject":"Remove inconsistent no_incremental option","message":"Remove inconsistent no_incremental option\n\nThis removes the \"--no-incremental\" flag and \"no_incremental\" option\nand uses the inverse of the existing \"--incremental\" flag or\n\"incremental\" option respectively instead.\n\nIt fixes the issue that \"incremental\" defaulted to False while\n\"no_incremental\" defaulted to None (effectively False) as well and both\noptions not being coupled so conflicting values were not caught. Parts\nof the code used one option while other parts used the other ones,\nleading to inconsistent behavior.\n\nCloses-Bug: #2138294\n\nChange-Id: I38d4f89fd1700a40d83ceb28ebd644a6aa1095a8\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a55c162ea8c96903850f3eda95ce79ace240e7b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a55c162ea8c96903850f3eda95ce79ace240e7b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
