)]}'
{"id":"openstack%2Ffreezer~973400","triplet_id":"openstack%2Ffreezer~master~Id19c290796cd107547020ae1bd6ca0426935ce7a","project":"openstack/freezer","branch":"master","topic":"fix/nova-cinder-glance-integration","attention_set":{},"removed_from_attention_set":{"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-01-16 14:20:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id19c290796cd107547020ae1bd6ca0426935ce7a","subject":"Fix freezer-agent modes and engines for Nova, Cinder and Glance","status":"ABANDONED","created":"2026-01-14 15:12:38.000000000","updated":"2026-01-16 14:20:54.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c2e7b511c38fd51ee1060156ecb08b5bdfdcaf54","_number":973400,"virtual_id_number":973400,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-15 18:05:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 15:53:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9771e5038b9a71d46fae9ccef0aa35c5bd08b4d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-14 15:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0094f6bcb495b352343c8ffa0fa90cebfbc02f9","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-14 15:18:30.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b882984c44ddd2a2680cfcbeb4618c88b337832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 15:53:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50fed4d97e884853ba156b0bc6a96af7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d583d157a1f040358018d7144594058a : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cdff09810e24a83b3025c443123e4e7 : FAILURE in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbe4742bdcac45ddb6080f44cac154ef : FAILURE in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c379fe1b2af44a5aab183d9eda14cf1 : FAILURE in 6m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/732b044401694b5885e62d480ea7d867 : SUCCESS in 3m 10s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/26cd72cce96440d7bd65312aa5fe76ba : SUCCESS in 29m 20s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/2d4cd1702fc845878aeaf4d3598ba340 : SUCCESS in 35m 13s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2aa65bcccfa4b329c91da0f27128dc4 : SUCCESS in 33m 51s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/259015008816485c8e2c20f861841bd9 : SUCCESS in 22m 13s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/63dfd03a515549378e04e869df4682d9 : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"281921315ddf0f10a48e884e894b7f68371ddf82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-15 15:55:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"18f242536265a2753fb938515c2faab52fe8855b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:39:12.000000000","message":"Patch Set 2: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c28cc3fc940c428faf2d707e42d45643\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a4227ed8ef42758bb832dfa2eee3b7 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c72c083a13469d988f718f3db72587 : FAILURE in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4362538182024a2d90689d0839ac2756 : FAILURE in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14d89e8ed3cc4b0899ce262e8848e3c9 : FAILURE in 3m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f831419d0898420e8e88ccb8f81a73dc : SUCCESS in 2m 28s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/514edd54881a4fdd92d4b06f25e115f0 : SUCCESS in 17m 37s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/a250e3feee244bab803562738f099caf : SUCCESS in 17m 46s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f11de6ad41745a49a461ecdc802bdc5 : SUCCESS in 38m 05s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/65c097b03b284cd88455834a8ab41c3d : SUCCESS in 31m 48s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/da41f751af014a77aa188c0bc2fa955d : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"8a2aab270a24d8cf6f768a13af69b1d68769bf2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-15 17:28:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"65f19a266675f12111f7263b7a5bec120e0ec8ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 18:05:38.000000000","message":"Patch Set 3: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19533970b0dc43c6bee412210cb8c477\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04ea226de89a4e7fb39aa30a80af2c65 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626620ae2dce466bb1feb6aebea36a1f : FAILURE in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3aa35abe6a314d1eae62163a87e2a14b : FAILURE in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20e4823a680b40d3a9213f05b1725174 : FAILURE in 8m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9fa942305bcf4ea792ea52dd076870a4 : SUCCESS in 3m 56s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/959f2be6ca9d4b198c89081f9e300d1b : SUCCESS in 34m 45s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/082d35a811ec45ef9f87a6a92e38fc74 : SUCCESS in 28m 52s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0feb64c92b70472485d9b442b8e4a03b : SUCCESS in 26m 34s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/b0b32e56df0e47f3ac0432bcb76d5eb5 : SUCCESS in 35m 26s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/e65cdc6b59364c6386062f1af7afbe25 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c2e7b511c38fd51ee1060156ecb08b5bdfdcaf54","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:20:54.000000000","message":"Abandoned\n\nAbandoning in favor of split up patchset successors:\n\n* https://review.opendev.org/c/openstack/freezer/+/973620\n* https://review.opendev.org/c/openstack/freezer/+/973623\n* https://review.opendev.org/c/openstack/freezer/+/973627\n* https://review.opendev.org/c/openstack/freezer/+/973625","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"167b660b2d4e3c410f47c2adba5aa02d1d022347","revisions":{"19e397d024fa3ba02f22bf854e02fadc236082b8":{"kind":"REWORK","_number":1,"created":"2026-01-14 15:12:38.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/00/973400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/00/973400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/00/973400/1"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-14 15:12:07.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-14 15:12:36.000000000","tz":60},"subject":"Fix freezer-agent modes and engines for Nova, Cinder and Glance","message":"Fix freezer-agent modes and engines for Nova, Cinder and Glance\n\nFor recent releases of Nova, Cinder and Glance, their modes/engines in\nfreezer-agent were failing due to outdated client, SDK or API calls.\nThis replaces most clients with the OpenStack SDK, migrates the code to the\ncorresponding SDK functions and fixes the workflows for nova, glance, cinder\nand cindernative modes/engines for both backup and restore.\n\nAlso includes some fixes for Python 3.x regressions.\n\nChange-Id: Id19c290796cd107547020ae1bd6ca0426935ce7a\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/19e397d024fa3ba02f22bf854e02fadc236082b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/19e397d024fa3ba02f22bf854e02fadc236082b8"}]},"branch":"refs/heads/master"},"dba3d45e0eab4f1b37f17976cdf8531bb48f7cbe":{"kind":"REWORK","_number":2,"created":"2026-01-15 15:55:45.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/00/973400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/00/973400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/00/973400/2"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-14 15:12:07.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-15 15:55:34.000000000","tz":60},"subject":"Fix freezer-agent modes and engines for Nova, Cinder and Glance","message":"Fix freezer-agent modes and engines for Nova, Cinder and Glance\n\nFor recent releases of Nova, Cinder and Glance, their modes/engines in\nfreezer-agent were failing due to outdated client, SDK or API calls.\nThis replaces most clients with the OpenStack SDK, migrates the code\nto the corresponding SDK functions and fixes the workflows for nova,\nglance, cinder and cindernative modes/engines for both backup and\nrestore.\n\nAlso includes some fixes for Python 3.x regressions.\n\nChange-Id: Id19c290796cd107547020ae1bd6ca0426935ce7a\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/dba3d45e0eab4f1b37f17976cdf8531bb48f7cbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/dba3d45e0eab4f1b37f17976cdf8531bb48f7cbe"}]},"branch":"refs/heads/master"},"167b660b2d4e3c410f47c2adba5aa02d1d022347":{"kind":"REWORK","_number":3,"created":"2026-01-15 17:28:22.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/00/973400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/00/973400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/00/973400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/00/973400/3"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-14 15:12:07.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-15 17:28:14.000000000","tz":60},"subject":"Fix freezer-agent modes and engines for Nova, Cinder and Glance","message":"Fix freezer-agent modes and engines for Nova, Cinder and Glance\n\nFor recent releases of Nova, Cinder and Glance, their modes/engines in\nfreezer-agent were failing due to outdated client, SDK or API calls.\nThis replaces most clients with the OpenStack SDK, migrates the code\nto the corresponding SDK functions and fixes the workflows for nova,\nglance, cinder and cindernative modes/engines for both backup and\nrestore.\n\nAlso includes some fixes for Python 3.x regressions.\n\nChange-Id: Id19c290796cd107547020ae1bd6ca0426935ce7a\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/167b660b2d4e3c410f47c2adba5aa02d1d022347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/167b660b2d4e3c410f47c2adba5aa02d1d022347"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
