)]}'
{"id":"openstack%2Ffreezer~973615","triplet_id":"openstack%2Ffreezer~master~Ica63643e14a554aff131b64bfeabeb6d53466112","project":"openstack/freezer","branch":"master","topic":"default-keystone-auth-url-version","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-19 11:57:00.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-16 15:32:33.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28271":{"account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"last_update":"2026-03-19 12:37:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica63643e14a554aff131b64bfeabeb6d53466112","subject":"Default to identity version 3","status":"MERGED","created":"2026-01-16 13:25:05.000000000","updated":"2026-03-19 12:39:08.000000000","submitted":"2026-03-19 12:37:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"973615-default-keystone-auth-url-version","meta_rev_id":"a55a033cf79cc50cab3a1f62d9ac060c0148db1c","_number":973615,"virtual_id_number":973615,"owner":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 12:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},"all":[{"value":1,"date":"2026-03-16 15:31:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},{"value":2,"date":"2026-03-19 11:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-16 15:32:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},{"value":1,"date":"2026-03-19 11:57:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 14:04:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-18 19:57:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-19 13:00:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-19 13:01:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-03-16 15:31:33.000000000","updated_by":{"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},"reviewer":{"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},"state":"REVIEWER"}],"messages":[{"id":"bbbd3ad81cab572555bb0e61fdc318d40ebaf290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2026-01-16 13:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6990358e40ca6d439e563fc1d0b497fd3b2bcaf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f59d9823054464a3699b7445a73d73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf49612c66543adaa1293d912117065 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728a6cdf302f4e5290bd9be451e89ede : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9745735a9d2344a186aedeea473f8165 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/991bdaace2924f7589d371deb27ff03c : SUCCESS in 4m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fda1aadcd2c43e09115683ce4faadc9 : SUCCESS in 4m 10s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/3f9af6b51b6348a1b653b6c79c471269 : SUCCESS in 35m 23s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/52beeb48ebab46368344384c73dda0c8 : SUCCESS in 29m 09s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e51b35b2ebdd45d086121925efcb109c : SUCCESS in 34m 44s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/475848c45a6f473d838684d35dfea087 : SUCCESS in 25m 03s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8bdf9789b8a46ce9cf2d9cd2d41e5e5 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":1},{"id":"56525055d52e5e05874122a5584ba0e2c11d5f7b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 19:57:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e2a3b5080fc5c8947b2d1b88b91c688fd91de72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2026-01-19 08:56:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2fb0b33523e3978b80e3eb06bf03c7b7e6017f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 09:38:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46a8d1739da04c54983b35dac5a3c160\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439f88d275894351a20a6a5bf7319416 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec43ff7a2004c698662e22e54db64c2 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45aac4036c674e7f85b908d0ea5c5227 : SUCCESS in 7m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/602461517c5644c386ab988bea872bb4 : RETRY_LIMIT in 1m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/304840713fdf45c18f491db9f361c535 : SUCCESS in 3m 45s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/196b0ab47298448db9153095866635aa : SUCCESS in 32m 25s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/3b072ead09ce4fa4b999bc4aef8cd238 : SUCCESS in 28m 32s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06521daab40a40348c29543c28a874ae : SUCCESS in 35m 05s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/94a8852383514062b60602d17b56cef1 : SUCCESS in 37m 46s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/181621490b57478d9891d0b0825634e8 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f2499e105f1776218a967b81b7b409e9cd18d62e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 09:39:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"282f01413f483fa0ee2f2af15066238442fda122","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2026-01-19 09:51:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6168cc5190aa95c2bc90496dbf587ebc230ebeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:33:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5937b70547884dbfa064e18998777c52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49b63698c6140cc99eb3a1b203529eb : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/847220e80a15431e935b02f54ea62e72 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70f7550ec2e3421090c610b47d8346c0 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/051ce2d9694346faaaa7a40a0e3cfb94 : RETRY_LIMIT in 1m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/935c3cef838f4f7a991b871232318184 : SUCCESS in 3m 33s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/c81dd5019c4442a984c881b6460acd2c : SUCCESS in 28m 08s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/19b79bef80bf4565bc33681d668a7453 : SUCCESS in 27m 41s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f884dd586c6a402fb00169eb09c232b6 : SUCCESS in 37m 21s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/ddfa4f8de1a44598b28af6fa03b8b6f4 : SUCCESS in 30m 17s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/052daa31e86c4a34b2e345cc5d4b102c : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"51c1b34a6719283d2d72b15f3c1a993a0d443ed6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 13:00:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02049bb7a20136e8b96957e61cba3ce48501e238","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 13:01:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9be802180a119df090d09083e7afd0aaf72b86b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:47:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c134ed588cf34fc5bf6a77c3780b4766\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4ecc26a0a34fdc9bfce97c705f81eb : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786286e49bd64547b3ada90ca9d2b0c7 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/002e37b5adde4e40bc232e2cb91a5554 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5e27acab5cc439d954feff40a58f60b : SUCCESS in 2m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f7b8aaf74a040a1b5c11104212c132a : SUCCESS in 5m 33s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/6f6b3ae8901649a98854eaafbd7760d1 : SUCCESS in 34m 04s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/43670a3dbbd84f809cb990aeb49de482 : SUCCESS in 36m 15s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f44a75347a6747a0ada6d3939b184d21 : SUCCESS in 33m 24s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c3c00cac86fc4784a095f7fba2941ff0 : SUCCESS in 41m 23s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1d66e79152044de8feeca2753084062 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1d0f02a395583a19e0480a4f153933473e494d8e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 15:10:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5a9db9c03691c2d6acf35f887b8387ffa37632f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2026-01-20 12:59:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2fb8bc85c96a9161fee4598dc41d0f52d5d802c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 13:26:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ca2b17205064c87b684558e97a2d6fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae914127722e484d9e906c910a502895 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63ed308d8b24e35952341b5ecc2b05f : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/253a6e7aa8644c30ba07c936aa496879 : FAILURE in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb255ecf15054ad7a5c81c494e2f03c0 : FAILURE in 2m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f2f63d9ace76492ead87d1260dcefbf8 : SUCCESS in 3m 46s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/c4c1fd162423448487e6777eb8f50506 : SUCCESS in 25m 01s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/649bb28ac5e94e0c8fce4a9964875a2e : SUCCESS in 16m 29s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95587fe0b0ef49d38ff32d16e287f8cd : SUCCESS in 21m 32s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/8c214f0a3d444df6b4ea8adf15ea1595 : SUCCESS in 17m 24s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/86d81e266ce94acabae90403286c7a55 : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":3},{"id":"8756191dd6a5e5c504d06afead1a6c64c02be0a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2026-01-20 13:46:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7dcc82e68fcc24e10ea5da1d29ba79761294c2e4","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2026-01-20 13:48:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4119f035f97fdf0e5c51830ffc60ee700beba949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 14:26:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29932dcfb49d45ac98cb08ee69492a31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b10b84b0692469caaf2fa2dc5227a08 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2812717ce4d14846bd2aaf4c1d6ecc03 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6341bbf7cf34daf90c92499c49476a1 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3d604cf491e4a8eb5993d8a54f2cc88 : SUCCESS in 3m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9de5aaea6d048469dd8a89b0162ffee : SUCCESS in 3m 44s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a302c5b31a06482b8d8453768c3cb675 : SUCCESS in 22m 55s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/fef1ddd5b3de459cb9d9ee91162f2bfc : SUCCESS in 14m 27s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7959069c15c04590856bee9ac8ac5595 : SUCCESS in 33m 24s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/91b0ac0e390d47c1bb44c75946dd6d79 : SUCCESS in 23m 31s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4207ce5dcee4874bd3f10b68c70f9f6 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":4},{"id":"c8f2d83ea85cc5e7b8489eba5d1e69480a5067e9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 18:33:15.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fc570e2b6a800cebeae6b2ed1b19bdb416599042","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-20 18:33:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"42d2f7518ccc20467c46aa84370c710cc9c5eac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2026-01-21 08:28:00.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"755fc6d792a52958aa973f319ffbd878d23f1a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 09:15:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2ea5b24fe4845ff88d50e021f013760\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca74f746ab3d40669bfff1aae780ac7e : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f831c4f554045a580a3b8000c730447 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7014143363f841009733bd150a1eb350 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09576a34f4e948a0bf5bfd0666c80e91 : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/765453320ab54472aa34337e8de631b6 : SUCCESS in 3m 29s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/61029d6688a548ccb05d71f581ea2c50 : SUCCESS in 22m 53s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/01f634dd243d4142a03bbc5707e2214b : SUCCESS in 32m 45s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e00be9a1d76d42bcb15ee0793239fece : SUCCESS in 17m 13s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/9f0f0eb33d094f9a9e2ca591f7c6dda7 : SUCCESS in 36m 32s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f56d1ec85b3e431286e421a79d6bf40b : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":5},{"id":"c58181a343f3f43e48b98b0afea24cc147e4ff4b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 14:27:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1ad3c83fa3922d5fc181fbe1beb0d9580b58803d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 14:28:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a1479940980deac6e986de9abba5d83b74c9ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 10:51:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"dc227379c0a6120dba8535bcf6d126b0947ab630","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 10:54:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d0fb2c8a35eaf701814e761939e12099e35c40b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 11:14:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"04f359ab295781cb2410677c444f5e24735d0a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:02:25.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe133091cadb4706bc64f49468cf510b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e9a22bf0524e2abd8dacf84d7043ce : SUCCESS in 2m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fda93ce1b364dbca48b1b1d3d9899f0 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8047a02768cc455c86a36c597686f33f : FAILURE in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a977762772ab49e5b5a9654b6560014f : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8af68863a13491a8eae8fa7b8a52e25 : SUCCESS in 6m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85c5355176c742dcb7ebcd576371c518 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/94f513c22dd445faa0c931391ee18324 : SUCCESS in 4m 51s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/78df018409034f7eb5ecf8c9ea99b199 : SUCCESS in 35m 43s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/1fc29ae7465d448faac209ef0d93aab3 : SUCCESS in 18m 50s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff3fd761d9b34989af853a8b56542692 : SUCCESS in 38m 46s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/12659eef8ef741abba139bf444039aff : SUCCESS in 41m 09s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7562d56512d45b189135079151f398c : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":7},{"id":"58ea6d67e54cf9467e5d77a8d60ac789da92e561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-13 14:39:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8050f10cae29f633e34c71ee61be6d23b7b3b17e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 15:24:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0994f7f5c1d54deaa7dc7cf7bcce65e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc83f9099b8442988da2a0ada34cc52a : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65a72845c19f4886aa29aac8085dabaa : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14d17cb4476545e984d8ba6d4c766306 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/524a7069ff15409dba427515591e401a : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b3a18ece6534b8bab5eaa9f710b94c3 : SUCCESS in 2m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73bdda49fd95454ba905d429d536a16f : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ec9710d01ed43dbad8d90ae8ba9e916 : SUCCESS in 3m 48s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/0565bc939e9f49ae86a09c5bf221fa40 : SUCCESS in 34m 23s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/e3ceb75e39b74d29be4c05f5ee2adf4c : SUCCESS in 32m 54s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77bf2709ff5a4ddba02dedf26d04af65 : SUCCESS in 35m 47s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/eeec0a1d6847470ea35745193db8d577 : SUCCESS in 38m 36s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba45cc675e3442a0910dc9bfbdb53bd3 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":8},{"id":"846d76656d01f579cd3346bef4ce10cbbb6b92af","author":{"_account_id":37587,"name":"Volodymyr Mevsha","email":"volodymyr.mevsha@gmail.com","username":"waerinfu"},"date":"2026-03-16 15:31:33.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"cb389919dc5d06b3858c86d81f742c595c140339","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-16 15:32:33.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dda7239b6643cc17e4f6d66fd40e9e54972d5870","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-19 11:57:00.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"90569d82eaa13e358a5b4d0320f71b819a12c18c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:57:32.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"30a9c5fb7dfbb1dd2030e24eec1f2dfd3a6ed75a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 12:37:30.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65c122e9faf943d79954146b23e49192\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c78027230614d22b04e268b6e3f03b3 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2457a86820544a3a83e1841297195dda : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce5f09c15f1468ea2b857e31a1900b0 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d5b2d9f5abc46659960b0edc4affed8 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d49228f17ec84d799c1258fdf5ddd9c8 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a70cb56526c6419ca7143b330ab6373c : SUCCESS in 4m 41s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/ee858414d3ee4881b30acd22547de921 : SUCCESS in 36m 46s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/27bc33890112467a84bd7ec37acdbdd8 : SUCCESS in 30m 27s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3c1210ac483422fb539fddd5507c0c9 : SUCCESS in 34m 40s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/bb462081974b4cea8dc0d1002bd2583c : SUCCESS in 35m 36s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/dccc3f3c0ebd4e4ca24d78ab2dd9dca6 : SUCCESS in 11m 49s","accounts_in_message":[],"_revision_number":8},{"id":"8bc109524ccf3c8288eea075c96d6e188a5af814","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 12:37:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"a55a033cf79cc50cab3a1f62d9ac060c0148db1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 12:39:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3c65465a5ae488abe2c7c737eeb9148\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/effc658e23bb4c5cbc37d05a102b67df : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad03196baf1142b798c6d031b4d4616b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b166a3b48b5ade755a835d6ba4ce75ba8b6687e7","revisions":{"b31edcc692d9321b9d8720c7151af9cdde1e41f7":{"kind":"REWORK","_number":1,"created":"2026-01-16 13:25:05.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/15/973615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/1"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b31edcc692d9321b9d8720c7151af9cdde1e41f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b31edcc692d9321b9d8720c7151af9cdde1e41f7"}]},"branch":"refs/heads/master"},"c03722ebbaa1a01173ca405706da064b65b47d06":{"kind":"REWORK","_number":2,"created":"2026-01-19 08:56:12.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/15/973615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/2"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-19 08:55:24.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\nAdditionally support for identity version 2 is dropped.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c03722ebbaa1a01173ca405706da064b65b47d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c03722ebbaa1a01173ca405706da064b65b47d06"}]},"branch":"refs/heads/master"},"b425abe9a95fdc930f7af06c2fbff0f93371502c":{"kind":"REWORK","_number":3,"created":"2026-01-20 12:59:53.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/15/973615/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/3"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-20 12:59:27.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\nAdditionally support for identity version 2 is dropped.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b425abe9a95fdc930f7af06c2fbff0f93371502c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b425abe9a95fdc930f7af06c2fbff0f93371502c"}]},"branch":"refs/heads/master"},"9c5f90622f285a6c0374c230ef94a81079c066d6":{"kind":"REWORK","_number":4,"created":"2026-01-20 13:46:39.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/15/973615/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/4"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-20 13:46:19.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\nAdditionally support for identity version 2 is dropped.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9c5f90622f285a6c0374c230ef94a81079c066d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/9c5f90622f285a6c0374c230ef94a81079c066d6"}]},"branch":"refs/heads/master"},"92158d8efa65ab279c40abcd5516fdef3b1c4a5a":{"kind":"REWORK","_number":5,"created":"2026-01-21 08:28:00.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/15/973615/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/5"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-21 08:27:43.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\nAdditionally support for identity version 2 is dropped.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/92158d8efa65ab279c40abcd5516fdef3b1c4a5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/92158d8efa65ab279c40abcd5516fdef3b1c4a5a"}]},"branch":"refs/heads/master"},"584535d8bdad157372070b8f7b960133271875fa":{"kind":"REWORK","_number":6,"created":"2026-03-13 10:51:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/973615/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/6"}}},"commit":{"parents":[{"commit":"b9cad5a6b60c424250b205e62655a14a9a4d7180","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9cad5a6b60c424250b205e62655a14a9a4d7180"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 10:51:33.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\nAdditionally support for identity version 2 is dropped.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/584535d8bdad157372070b8f7b960133271875fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/584535d8bdad157372070b8f7b960133271875fa"}]},"branch":"refs/heads/master"},"89cc77f62f26674208d56c4ad4f067dfd4ca57ab":{"kind":"REWORK","_number":7,"created":"2026-03-13 11:14:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/973615/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/7"}}},"commit":{"parents":[{"commit":"b9cad5a6b60c424250b205e62655a14a9a4d7180","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9cad5a6b60c424250b205e62655a14a9a4d7180"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 11:14:22.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\nAdditionally support for identity version 2 is dropped.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/89cc77f62f26674208d56c4ad4f067dfd4ca57ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/89cc77f62f26674208d56c4ad4f067dfd4ca57ab"}]},"branch":"refs/heads/master"},"b166a3b48b5ade755a835d6ba4ce75ba8b6687e7":{"kind":"REWORK","_number":8,"created":"2026-03-13 14:39:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/973615/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/15/973615/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/15/973615/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/15/973615/8"}}},"commit":{"parents":[{"commit":"b9cad5a6b60c424250b205e62655a14a9a4d7180","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b9cad5a6b60c424250b205e62655a14a9a4d7180"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2026-01-16 13:22:40.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-13 14:39:11.000000000","tz":60},"subject":"Default to identity version 3","message":"Default to identity version 3\n\nWhen there is no identity version given and no version specified\nin the AUTH_URL, this patch now defaults to identity version 3 and\nadjusts the AUTH_URL internally instead of raising an exception.\nAdditionally support for identity version 2 is dropped.\n\nChange-Id: Ica63643e14a554aff131b64bfeabeb6d53466112\nSigned-off-by: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b166a3b48b5ade755a835d6ba4ce75ba8b6687e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/b166a3b48b5ade755a835d6ba4ce75ba8b6687e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
