)]}'
{"id":"openstack%2Ffreezer~973620","triplet_id":"openstack%2Ffreezer~master~I9f054fe6b25865a1d703b6d2daf507a59cd5d319","project":"openstack/freezer","branch":"master","topic":"fix/nova-cinder-glance-integration","attention_set":{},"removed_from_attention_set":{"36848":{"account":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"last_update":"2026-01-26 15:38:42.000000000","reason":"Change was submitted"},"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-01-26 15:38:42.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-19 13:12:08.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I9f054fe6b25865a1d703b6d2daf507a59cd5d319","subject":"Fix issues in Python code","status":"MERGED","created":"2026-01-16 13:44:28.000000000","updated":"2026-01-26 15:39:32.000000000","submitted":"2026-01-26 15:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973620-fix/nova-cinder-glance-integration","meta_rev_id":"b573e3f62b6ba5dfa7821ca4fa9313056515ced9","_number":973620,"virtual_id_number":973620,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 15:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-01-19 13:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-19 19:50:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-01-23 08:05:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 14:24:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-18 19:58:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-19 07:52:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-01-19 13:01:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-01-19 19:50:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-19 19:50:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36848,"name":"Alvaro Soto","display_name":"Alvaro Soto","email":"alsotoes@gmail.com","username":"alsotoes"},"state":"REVIEWER"}],"messages":[{"id":"d1ed7c0bc398b4863861b7bb3620ecb4535da095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 13:44:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99ae5550d0d0712f360ab8bd3969deda1f883b52","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:01:44.000000000","message":"Topic fix/python-fixes removed","accounts_in_message":[],"_revision_number":1},{"id":"c7788c32704dec5044c48317903944f9f95dce17","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 14:01:50.000000000","message":"Topic set to fix/nova-cinder-glance-integration","accounts_in_message":[],"_revision_number":1},{"id":"55e7327657e953db8787a7e53c0cad40b1bd81d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:24:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6272d995945c42be8e69b132c48c6143\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd3b20d6e31a470f84e5582e818fd7d6 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d96ffe4d044cd7b62536b5ecb2cd2c : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3a94f375a794f819941cad90641ea23 : FAILURE in 12m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0844b428b35c44b0aaea10b690359ad0 : FAILURE in 17m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78ba79ebeed84c3c9dc5cd4ef8e7ad2e : SUCCESS in 2m 47s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/bb71676e7eb947088f4133e22af7696b : SUCCESS in 34m 15s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/5537e398069040009f0a94c6df50ddda : SUCCESS in 33m 01s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23c04ef56acd40c79e1c802d43d9fcf3 : SUCCESS in 35m 03s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/d756526a3bf74d8c9fe0d4a03e92655f : SUCCESS in 26m 36s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f076ed3f5d31493c8b4ee0358c17ee2a : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"90532a1bef68e81ec342682ff756b591eb52c8e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 14:48:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_27665\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}],"_revision_number":2},{"id":"bb414ccbf59e0b04699aebd9e4ac80330d3995fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:48:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f33971adf04ca18d9a01591cbf4bf6\n\n\nWarning:\n  Dependency cycle detected and project openstack/freezer doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":2},{"id":"ec7223ba1d70cf73d19c964006db845c1664e907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-16 14:49:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_27665\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}],"_revision_number":3},{"id":"bcd09a4914530c74ac4200ea13157d8049eba184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 15:01:23.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"71bce5be26a3d354f195a82a0b7aee3c5232d407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 15:41:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b397bede94f6471395b99c29155d375c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1584d38753f412ab2b43baa6cf112fa : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98349e71fad4c4ea31cd9ac0a771be3 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3669cf80209847b097af742ead03605d : FAILURE in 12m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/514966db586f465c9d27871126c37862 : FAILURE in 16m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4629d0a1e03a48eca709e7e8ca424dd2 : SUCCESS in 4m 17s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/91419f098d7d41209e3251848394c94e : SUCCESS in 33m 28s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/4e143e5ee2a04185b724df92e086c211 : SUCCESS in 34m 38s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b97b1ca11c34c0cb156e8e034bc595d : SUCCESS in 34m 49s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/79a2cdac6fc94e83891850061fcd79bc : SUCCESS in 37m 11s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/71af7090e86b41cbaafa90381ab5ff09 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c618c8699590b6f78caafc727210d862bea37da2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 16:24:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7a235ef614d71e49ea91327de5c5af59a843a0eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:07:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e39e9f1b83464c28a9da1eec638a4446\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2bb54c5ea44711bd41adfd6d434279 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c65b071ad54324a2a78532d121a404 : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ff9761e450440fe934e8d47541a4c9f : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2baca0d83934b06a2803f8eb93b1d6d : SUCCESS in 6m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/af84dc1c3b8d423aba5881d9d1d6cada : SUCCESS in 4m 21s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/68e7643bf9564201bc098226677c03fe : SUCCESS in 22m 29s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/d0a20c625f73492f993ee3a9e93894c7 : SUCCESS in 17m 35s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d15847869ea45aeb3d741ee658fdc93 : SUCCESS in 37m 16s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/7fa961901a2f49c0b0b0a14dd3048839 : SUCCESS in 38m 01s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/53c91db63d694f7ba1846f54243e53f6 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":5},{"id":"edcab482bbeac1c28ad4763cfc506cf59423678c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-01-16 17:13:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"54f7989802301dfe97214d906f56dd1eab9a931e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:02:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13fc2dd248ce4977a113b067905787a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/712cdeee73d846b1993bc8134f80ecba : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57936b8a04f43c5b1cd7291d1f2968c : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/632e12cb52ed4dc19248df62706af9db : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18c1753a67f54d6fbb120a50936ff0db : RETRY_LIMIT in 1m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0767ef4f19b4442aad608228c0651ab0 : SUCCESS in 4m 39s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/7adfe5f81c1147c2a530963716c57a67 : SUCCESS in 36m 01s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/03ffc93debe047329b81ab06c4121c3f : SUCCESS in 38m 00s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adfef5a4dfac4182b500648aea1bce19 : SUCCESS in 41m 08s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/7a42b5684a0f4b6890a3d923b0dc56bd : SUCCESS in 21m 20s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/b60251a86fa54a429abf5d2fcc38c4b8 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":6},{"id":"1885dde48bbe880e34bf8288c0ad8cf0e988a8b1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 19:58:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"46d77886974c199c9b4c0c2ab7079f0c7aa7b299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-18 20:40:57.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a192c33b6c84a62b08951f42b50f99c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea1075bebe4484a9c8217b4dd969707 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ceee1864614ea0b6b15ddd49f7e4a6 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2afb5cc2cfad44a496d7a502866fb2d4 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4e442c2487a4c03b1d0f12dbd6d8dd8 : RETRY_LIMIT in 1m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7c8eaad670f4a98bed3c7bdc480d2fe : SUCCESS in 2m 36s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/423c666233864d18939a9da4ab050d2f : SUCCESS in 37m 26s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/4472ba49b344467e9cc74a97aa85828e : SUCCESS in 33m 18s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d68c1d09b724849bbd685305d9a5271 : SUCCESS in 21m 44s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c54dde42bb3249c8bbddd0a6e9cb19f3 : SUCCESS in 24m 52s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0bcface860d4a24b0669ad8bbc46707 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":6},{"id":"c1f1216e18a56a9805da4cb39e76fd8706195b50","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 07:52:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"40e61f26b9f7a44bc4b497a56950534485764a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 08:34:42.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca633c7be8054a3ebe5aae992f8e6191\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e432be77ddc49d6a423cd6c7ff5d541 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab90e3c0fc5c4133b7962140630f5cd4 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0454f131f48240d6bba21a56cd7bacd9 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c729ad950c1442f8de8e5da425324e9 : RETRY_LIMIT in 1m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6452113d57d4ad1983bf1e150ad0827 : SUCCESS in 4m 59s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/73baeb4e6560418881f079f4cc8c30eb : SUCCESS in 36m 37s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/8f936b0197994b038aa52bd6c04096c5 : SUCCESS in 26m 04s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67e616ff62f840948354aa445b27c73b : SUCCESS in 35m 31s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/2503add0799c417a96c8dfdfa1d7c157 : SUCCESS in 37m 28s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/770c5f8311144e768fc00f5844aed973 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":6},{"id":"f23ee0c618e6cc50c2c86e1c08e4b5e4b048394b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 12:23:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0f31dd64b4f309b9480acf9dda144de2a6489be6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:45:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c42d1a248e4a4e1f89168c62181e0aa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45f0f9d642740feb99099160ca495c1 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3bbeeef7228463b83018760123b2aad : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05b8adb7715b4137a17e34f2977b1c51 : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d54647b781de459cb45e3e2408d519ec : SUCCESS in 2m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e9fbc243972f42619ee981e354c75dcf : SUCCESS in 3m 37s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/2129954a176a4147b7afe26c1a11078e : SUCCESS in 16m 49s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/ed52392a6e9a435483cb59563bff006b : SUCCESS in 16m 45s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab597c1d469b4cd786bf82f1264b7aae : SUCCESS in 17m 24s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/c45de5ed422745a880f8c488330f4cc3 : SUCCESS in 20m 43s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/edaacdc83e3340718a2abab572d5ed7b : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":6},{"id":"0bac16072e2fc244cb965654459643b6f2642233","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-19 13:12:08.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9545d86aab23103a1bf43ac72e1b6a10234f708f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:50:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"50aed2f5065203534b30670c1695e5139cf5e1db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-23 08:05:33.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4844982d00b9cd8e8380d65f08b592ae9efdfdf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 08:05:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ced3978d5a18efd62857a19dcd19bdf3623e7146","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 08:50:22.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1374c0cf43934cc59d95d6972a36f10a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38234623e31485e9e8517ed8557a2b5 : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b06a64bf29440592a2d9e6d197bd28 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45bece64e0c0485b9064350c83ad483e : SUCCESS in 2m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68633fe215d24278b481b985a62e820d : SUCCESS in 2m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59d3725ebb9e439d92b6396a8c6d9eaf : SUCCESS in 4m 41s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/a670f41bfca04b5fb406ef931c42e16a : SUCCESS in 39m 13s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/39cd63a5b5b447d08e20314438f89c07 : SUCCESS in 20m 13s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86a020dc796249cc8326ff64efeebd0f : FAILURE in 22m 16s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/9cb2365018b7402aa9b4d985c6a92b37 : FAILURE in 21m 41s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/686c4404e44f4d4c8cae7587aa3950fb : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":6},{"id":"fd25a5e93981ea33c200ecb15d906cae3a077033","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-26 14:08:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"58eb3affa44f9b5c6e889e17f9a7dca76cb49f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 14:52:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/435296140dee49f78db8e863b6505ed1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50b351e35ac445cb2f89706d9fd37ad : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7901f66c720d4b1a8282156e99e7386a : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26dc0f268c62479688a1456a6557f0b1 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f53d0fc539314495800cf0c831cc5919 : SUCCESS in 3m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5cad7636f06e4f849d3bc77ba7982359 : SUCCESS in 3m 28s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/11e232e0af19447e9faa9f6fa7aa52a6 : SUCCESS in 15m 25s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/2bda4175284e4bb2b0792eef280f3b9e : SUCCESS in 35m 19s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2df8d36a5fee42aaa10cb6a9ada28aa3 : SUCCESS in 19m 44s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/831ecac9052045308411c1e9b5462892 : SUCCESS in 39m 07s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc435fc5398845b389d76acb27ce2365 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":6},{"id":"79fb9e275e5858e6cbef32e3a9c3ae715500e4a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 14:53:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"879bdabcf585b2f0b7bc98ee1e86f55ad08c648d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:38:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1630f6f55149908a40f9de8478b981\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a9b9f954364003b713905b0b8c6213 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a88eda5aa6ff439a99b98ff6674d1399 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61f9ef5374d54bfba20eec8de6350aa0 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7da107a9852b4fedb4f22f86f1446492 : SUCCESS in 3m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35c66b6a93554fac8ef240fa45f0e3c9 : SUCCESS in 2m 16s\n- freezer-tempest-agent https://zuul.opendev.org/t/openstack/build/e734d56800d24606a0d8c90db88f1e3a : SUCCESS in 27m 41s\n- freezer-tempest-scheduler https://zuul.opendev.org/t/openstack/build/db27485fb1264ff9b02e3df33169f93f : SUCCESS in 30m 14s\n- freezer-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e71f608650794043a7144658767ed312 : SUCCESS in 36m 28s\n- freezer-ubuntu https://zuul.opendev.org/t/openstack/build/1f942c221797464eb0fae6f8cba8803e : SUCCESS in 40m 30s\n- freezer-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6779eae594b4ecba715c2f76107674f : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":6},{"id":"f238568f93652c6029c509d82a3d3fdf983d6383","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:38:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b573e3f62b6ba5dfa7821ca4fa9313056515ced9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:39:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc8f415b2b27496dba6d1e3ac5801b99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a875a3d994b1429bbc085eb4bdac284e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"74375ae6038dfe0be2dbe211fe75b68124c600c0","revisions":{"ddc50ec44e02f4d2fe8b795a54b2eef70dcbd3fd":{"kind":"REWORK","_number":1,"created":"2026-01-16 13:44:28.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/20/973620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/20/973620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/20/973620/1"}}},"commit":{"parents":[{"commit":"a2211c2773031930246f8c48d681e8947d6579bd","subject":"tox: remove testenv:py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/a2211c2773031930246f8c48d681e8947d6579bd"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:32:31.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:32:31.000000000","tz":60},"subject":"Fix issues in Python code","message":"Fix issues in Python code\n\n* Fixes JSON read/write operations which historically executed on\n  file handles opened in binary mode but need text mode now.\n* Fixes backup enumeration crashing when no timestamp is given so\n  that the latest backup is automatically selected.\n* Fixes ReSizeStream for recent Python versions to not raise\n  StopIteration anymore. [^1]\n\n[^1]: https://peps.python.org/pep-0479/#writing-backwards-and-forwards-compatible-code\n\nChange-Id: I9f054fe6b25865a1d703b6d2daf507a59cd5d319\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ddc50ec44e02f4d2fe8b795a54b2eef70dcbd3fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/ddc50ec44e02f4d2fe8b795a54b2eef70dcbd3fd"}]},"branch":"refs/heads/master"},"d365aeeb5904ee061e4a8d30c742c9cf53accece":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-16 14:48:56.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/973620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/20/973620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/20/973620/2"}}},"commit":{"parents":[{"commit":"d2cd3908279c5c1fa831d5a1aa49b9d5c1df2774","subject":"Fix Glance integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d2cd3908279c5c1fa831d5a1aa49b9d5c1df2774"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:32:31.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:48:56.000000000","tz":0},"subject":"Fix issues in Python code","message":"Fix issues in Python code\n\n* Fixes JSON read/write operations which historically executed on\n  file handles opened in binary mode but need text mode now.\n* Fixes backup enumeration crashing when no timestamp is given so\n  that the latest backup is automatically selected.\n* Fixes ReSizeStream for recent Python versions to not raise\n  StopIteration anymore. [^1]\n\n[^1]: https://peps.python.org/pep-0479/#writing-backwards-and-forwards-compatible-code\n\nChange-Id: I9f054fe6b25865a1d703b6d2daf507a59cd5d319\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d365aeeb5904ee061e4a8d30c742c9cf53accece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/d365aeeb5904ee061e4a8d30c742c9cf53accece"}]},"branch":"refs/heads/master","description":"Rebase"},"0c4cbcb3d3e4e74d623e7ed45122168a02739fca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-16 14:49:25.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/973620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/20/973620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/20/973620/3"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:32:31.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 14:49:25.000000000","tz":0},"subject":"Fix issues in Python code","message":"Fix issues in Python code\n\n* Fixes JSON read/write operations which historically executed on\n  file handles opened in binary mode but need text mode now.\n* Fixes backup enumeration crashing when no timestamp is given so\n  that the latest backup is automatically selected.\n* Fixes ReSizeStream for recent Python versions to not raise\n  StopIteration anymore. [^1]\n\n[^1]: https://peps.python.org/pep-0479/#writing-backwards-and-forwards-compatible-code\n\nChange-Id: I9f054fe6b25865a1d703b6d2daf507a59cd5d319\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0c4cbcb3d3e4e74d623e7ed45122168a02739fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/0c4cbcb3d3e4e74d623e7ed45122168a02739fca"}]},"branch":"refs/heads/master","description":"Rebase"},"8d57bc9bd120afd32b0b1e33b5faa7c7a08b807f":{"kind":"NO_CHANGE","_number":4,"created":"2026-01-16 15:01:23.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/20/973620/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/20/973620/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/20/973620/4"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:32:31.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 15:00:52.000000000","tz":60},"subject":"Fix issues in Python code","message":"Fix issues in Python code\n\n* Fixes JSON read/write operations which historically executed on\n  file handles opened in binary mode but need text mode now.\n* Fixes backup enumeration crashing when no timestamp is given so\n  that the latest backup is automatically selected.\n* Fixes ReSizeStream for recent Python versions to not raise\n  StopIteration anymore. [^1]\n\n[^1]: https://peps.python.org/pep-0479/#writing-backwards-and-forwards-compatible-code\n\nChange-Id: I9f054fe6b25865a1d703b6d2daf507a59cd5d319\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8d57bc9bd120afd32b0b1e33b5faa7c7a08b807f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/8d57bc9bd120afd32b0b1e33b5faa7c7a08b807f"}]},"branch":"refs/heads/master"},"e9c088782137f558cb566cf91e9899a2221509a5":{"kind":"REWORK","_number":5,"created":"2026-01-16 16:24:14.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/20/973620/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/20/973620/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/20/973620/5"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:32:31.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 16:23:57.000000000","tz":60},"subject":"Fix issues in Python code","message":"Fix issues in Python code\n\n* Fixes JSON read/write operations which historically executed on\n  file handles opened in binary mode but need text mode now.\n* Fixes backup enumeration crashing when no timestamp is given so\n  that the latest backup is automatically selected.\n\nChange-Id: I9f054fe6b25865a1d703b6d2daf507a59cd5d319\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e9c088782137f558cb566cf91e9899a2221509a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/e9c088782137f558cb566cf91e9899a2221509a5"}]},"branch":"refs/heads/master"},"74375ae6038dfe0be2dbe211fe75b68124c600c0":{"kind":"REWORK","_number":6,"created":"2026-01-16 17:13:11.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/20/973620/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/freezer","ref":"refs/changes/20/973620/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/freezer refs/changes/20/973620/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/freezer refs/changes/20/973620/6"}}},"commit":{"parents":[{"commit":"c5c6f95958e9601b31f8f64fc741e0c1db757dac","subject":"Fix wrong commands in glance backup examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/c5c6f95958e9601b31f8f64fc741e0c1db757dac"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 13:32:31.000000000","tz":60},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-01-16 17:13:00.000000000","tz":60},"subject":"Fix issues in Python code","message":"Fix issues in Python code\n\n* Fixes JSON read/write operations which historically executed on\n  file handles opened in binary mode but need text mode now.\n* Fixes backup enumeration crashing when no timestamp is given so\n  that the latest backup is automatically selected.\n\nChange-Id: I9f054fe6b25865a1d703b6d2daf507a59cd5d319\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/74375ae6038dfe0be2dbe211fe75b68124c600c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/freezer/commit/74375ae6038dfe0be2dbe211fe75b68124c600c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
